)]}'
{"id":"openstack%2Fpuppet-ironic~758296","triplet_id":"openstack%2Fpuppet-ironic~master~Ica7bcdc53cd3d9f8350a1b49b311070a26116841","project":"openstack/puppet-ironic","branch":"master","topic":"prep-release-wallaby-start","hashtags":[],"change_id":"Ica7bcdc53cd3d9f8350a1b49b311070a26116841","subject":"Bump version for the start of Wallaby","status":"MERGED","created":"2020-10-15 01:43:42.000000000","updated":"2020-10-15 14:14:12.000000000","submitted":"2020-10-15 14:14:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758296-1602771252231-a731ec30","meta_rev_id":"95d95164fb71f46f84b81112a5de99444a33462b","_number":758296,"virtual_id_number":758296,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 14:14:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-15 14:05:27.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-10-15 07:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-10-15 14:05:27.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 07:31:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-15 14:05:27.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-15 14:14:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d46bdf57dabe722f01077aae576e66b18a99d05e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-15 01:43:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74b97995d406519f2b63cd6a58cc40755cdd5cc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 01:52:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/f9a5e999c5ac4ea38e47baf6b5bc4a5c : SUCCESS in 2m 42s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5dd725ac75e44716958a5a62e8602c80 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"06d10ed25502fdc5f25db2d14abb8f0964e00662","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-15 07:31:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8d1c8c86ed9156cd21ee1dd447e5b0e18e7991b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-15 14:05:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40b27f1641e3048fe669cbb9d4cba26375539eab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:05:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc98ce822ff74927756c0cb2d5d33060a00b108e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:14:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"99af240b2ed30910621ed655275ba9c1d4e48ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:14:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/4b71d5585d8f42ee92d81fdc99699ce5 : SUCCESS in 2m 48s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d95916061a134e20889b288a56751968 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c962ee7ebed56a6ef0c246a0d35050824af0c99","revisions":{"8c962ee7ebed56a6ef0c246a0d35050824af0c99":{"kind":"REWORK","_number":1,"created":"2020-10-15 01:43:42.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/96/758296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/96/758296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/96/758296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/96/758296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/96/758296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/96/758296/1"}}},"commit":{"parents":[{"commit":"ef29d243b017dccb9f051da6f0eea6dfdc476864","subject":"Merge \"Cleanup codes for CentOS/RHEL7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/ef29d243b017dccb9f051da6f0eea6dfdc476864"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-15 01:42:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-15 01:42:49.000000000","tz":480},"subject":"Bump version for the start of Wallaby","message":"Bump version for the start of Wallaby\n\nBump the major version for the Wallaby release after Victoria has\nbeen branched.\n\nChange-Id: Ica7bcdc53cd3d9f8350a1b49b311070a26116841\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/8c962ee7ebed56a6ef0c246a0d35050824af0c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/8c962ee7ebed56a6ef0c246a0d35050824af0c99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
