)]}'
{"id":"openstack%2Fpuppet-ironic~758698","triplet_id":"openstack%2Fpuppet-ironic~master~I2fb901e22e01c5b3acda20ee286413c13ba9b90c","project":"openstack/puppet-ironic","branch":"master","topic":"at-params","hashtags":[],"change_id":"I2fb901e22e01c5b3acda20ee286413c13ba9b90c","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2020-10-19 03:41:18.000000000","updated":"2020-10-19 10:36:30.000000000","submitted":"2020-10-19 10:33:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758698-1603103633091-1d1e633c","meta_rev_id":"044b12d3fa7244df7a44ad863a997c58225a4997","_number":758698,"virtual_id_number":758698,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 10:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-19 08:43:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-10-19 07:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-19 08:43:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 07:52:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-19 08:43:01.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-10-19 10:33:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24ffe05bd7afaecd9982b38de5f56c81e7ab6f66","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-19 03:41:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22f370815a98d18397c40b58a93a53e42a4e1a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 05:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d84d8564da3d424eada1978d4409ac1e : SUCCESS in 4m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/683154d9441e4940a8d82507245636eb : SUCCESS in 4m 32s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/95da331b889d45f386850b5df1fda8c4 : SUCCESS in 13m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a64488a860040f09e0a2c697ebbd59a : SUCCESS in 11m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eddb2a60f96240de81e2b61d04597678 : SUCCESS in 14m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30be6c51c01042239dbba92b4d317788 : FAILURE in 28m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e33b507003304628807f6405525e0965 : SUCCESS in 1h 52m 43s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a669193d5280442c97f279c3d0a39570 : SUCCESS in 27m 12s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c37eee3ad2b94364a21b6b94528a9cd4 : SUCCESS in 15m 36s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f55ecc10b8564bc88b1d1eee38064eb6 : SUCCESS in 1h 31m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3efb97635efd471a8a64cc016c3e855b : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"85acbb41efdead94aff749f16afd1936236155e1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-19 07:52:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d3d4e490bddfa68fcd63d03a64b9385d9e05060","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-19 08:43:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95e568b12fbb05d7505959f5bf78011594609d08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 08:43:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fb60ad96167cf2f96de0ed5bf5fb592932c4a25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 10:33:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/860d85131b264d17890d686afd4ba525 : SUCCESS in 7m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9154df546bcd4f1a899b8e496d39c6d9 : SUCCESS in 7m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6f4a099acfd94aa6a09b5a5bd58e4d4e : SUCCESS in 17m 33s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a2c01a5471b4efdae15fc51f4e51f5a : SUCCESS in 15m 31s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7e288e22c174063812e37af4e9f011d : FAILURE in 32m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d5f5db20b47d46d0b63d7b9fc2896b4d : SUCCESS in 1h 43m 16s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0aedf1153fb84e198e13a7ee3ccfac0d : SUCCESS in 23m 18s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b63b3e099a6b4f1da2bc0a94d493fb9e : SUCCESS in 19m 11s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9455008b53574d46b08d58b0f7cb2fa2 : SUCCESS in 1h 38m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5bd242020f34d6ca56019f4d0374569 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":1},{"id":"158ef84ff7bb9a5af0f690d7d910d43b9bb63f43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 10:33:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4bd78c112b9f2235d6f6a03c45c7aed7e75f7b2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 10:36:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b524c3e30e334af5a5898f0a11f1994a : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f17339fb7d5e16102cebd5cf2d41b3ea084ec3d1","revisions":{"f17339fb7d5e16102cebd5cf2d41b3ea084ec3d1":{"kind":"REWORK","_number":1,"created":"2020-10-19 03:41:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/758698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/98/758698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/98/758698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/98/758698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/98/758698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/98/758698/1"}}},"commit":{"parents":[{"commit":"82a5587266d5b0d5234791dae043576429692c9e","subject":"Prepare Victoria RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/82a5587266d5b0d5234791dae043576429692c9e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-19 03:40:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-19 03:40:58.000000000","tz":540},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: I2fb901e22e01c5b3acda20ee286413c13ba9b90c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f17339fb7d5e16102cebd5cf2d41b3ea084ec3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f17339fb7d5e16102cebd5cf2d41b3ea084ec3d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
