)]}'
{"id":"openstack%2Fpuppet-ironic~759671","triplet_id":"openstack%2Fpuppet-ironic~master~If95ff0d0422142c3ed74eba4110969479a51139d","project":"openstack/puppet-ironic","branch":"master","topic":"redfish-firmware-update","hashtags":[],"change_id":"If95ff0d0422142c3ed74eba4110969479a51139d","subject":"Add support for the firmware update feature in redfish driver","status":"MERGED","created":"2020-10-26 11:55:30.000000000","updated":"2020-10-27 11:05:04.000000000","submitted":"2020-10-27 11:03:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759671-1603796626443-d9ea6aaa","meta_rev_id":"2aeb7d146b489db890f417cc22f6dcc37ca19ac9","_number":759671,"virtual_id_number":759671,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2020-10-26 14:38:48.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-27 11:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-27 09:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-27 08:46:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-27 09:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-27 08:46:27.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-10-27 09:29:16.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-10-27 11:03:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e263c94272860bb2c29c75c77e2ed614a5a0c603","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-26 11:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22deb64ccebafab403666f825c158a8aa8b6a9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 13:52:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01ee8e78c05d486692f47de82c004512 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d62a3a6eab14bec873256c4c1a1f172 : SUCCESS in 3m 11s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/89b17b1e016346dd9afc51cf2387f111 : SUCCESS in 16m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68c96b68c6254e04a590777ef4948786 : SUCCESS in 9m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb99db004c18448f93329fca203bcb81 : SUCCESS in 10m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74b4c99c709741c3807ec7b9896c4b82 : SUCCESS in 1h 03m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bfe870f725574c0496111b685e3d41ea : FAILURE in 1h 49m 59s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/628c963625ef43e58badb9f8128d204e : SUCCESS in 23m 22s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a0b8ee12f7741fd8c3c207c8839934c : SUCCESS in 19m 08s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/34467c72a8da4467946fecd1356174ca : SUCCESS in 1h 39m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2788839ed32e44738802a75d2a7211fd : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"af876f461a8df146b0f04113a40aeaf7f005e68f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-26 14:38:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a77549bd6d1a16f179fe31be5d339e91731f71df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 16:43:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7c278974e0e4de68fdeb82572063713 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/934bf6f362864a19849b2c9beb202525 : SUCCESS in 3m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/35a33a626cea46a4863b4a037e20ca91 : SUCCESS in 11m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/746fa2a22f424ad7b5fc80ffef806852 : SUCCESS in 16m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92abc3800d1b4efd88e88530082e4223 : SUCCESS in 16m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1f48a4bf154427bb092f28348e34189 : SUCCESS in 1h 13m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3ea661fc7dc14ddebbc9584f9a5a0c74 : SUCCESS in 1h 44m 59s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a8a5c4bbfa4147d4bb679826a5314de3 : SUCCESS in 27m 36s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22d7bf4ddda640c0b66cced802fb6035 : SUCCESS in 21m 18s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/eb1c10982ada4c2cb9ddb552cef1259c : SUCCESS in 1h 45m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0aab08d6e494b0d9299d0e0de246cee : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"82a8d10a5016fc04b4991f4b8189f5e3e7369d35","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-10-27 08:46:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01c623995f1fc4f19896a80acf5fd083269a77b3","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-27 09:29:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2bf4dd08d9011e64d07f0fd6d4b1b8dd72ff9fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 09:29:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65c4b4767b725dbab0c6805625715e02b398c4dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 11:03:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fd871f7ba6d4e4ca495afe46b187181 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afcee468f8ec4619a2048b6d998e2c56 : SUCCESS in 3m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cbebb02a0642480087be89e060268473 : SUCCESS in 12m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d95207f16e0d43ffb4d9bdb9389677c9 : SUCCESS in 9m 43s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfe22f77a29b47eabd73646233453376 : FAILURE in 6m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/56837f8eeff1433d9254378f8a2418e7 : SUCCESS in 1h 28m 32s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ec18c2544dc14800960f027597707128 : SUCCESS in 18m 08s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8121d2e499fc46f3a44087f829fd9c56 : SUCCESS in 13m 14s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/22092801e2754960b4534baaa173d01e : SUCCESS in 1h 26m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/388bab5140a544e2876b0302fa215492 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c2a1a37b19f005ff29eed5e75a6d2a9693b8573c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 11:03:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3c9c9b6df6c9c9d22c1ead770758b27ce084051","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 11:05:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/912f8ba809ea48cca18d5a8d38ac177e : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4de96545792b6ed6375ca7ce07b58f617a0f0e0b","revisions":{"4de96545792b6ed6375ca7ce07b58f617a0f0e0b":{"kind":"REWORK","_number":1,"created":"2020-10-26 11:55:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/759671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/71/759671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/71/759671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/71/759671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/71/759671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/71/759671/1"}}},"commit":{"parents":[{"commit":"02d0fea8296b59236a9d0d93ebdec8c9467dd8eb","subject":"Fix the misplaced release note file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/02d0fea8296b59236a9d0d93ebdec8c9467dd8eb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-26 11:54:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-26 11:54:09.000000000","tz":540},"subject":"Add support for the firmware update feature in redfish driver","message":"Add support for the firmware update feature in redfish driver\n\nThis patch introduces parameters related to the firmware update, which\nwas added to Ironic during Victoria cycle[1].\n\n[1] f0e0ef634e55f46eda4942e9665d35c70d305365\n\nChange-Id: If95ff0d0422142c3ed74eba4110969479a51139d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/4de96545792b6ed6375ca7ce07b58f617a0f0e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/4de96545792b6ed6375ca7ce07b58f617a0f0e0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
