)]}'
{"id":"openstack%2Fpuppet-ironic~763703","triplet_id":"openstack%2Fpuppet-ironic~master~I7c461005b2686e50dc8affd9bf2ecb0115654bec","project":"openstack/puppet-ironic","branch":"master","topic":"bug/1904962","hashtags":[],"change_id":"I7c461005b2686e50dc8affd9bf2ecb0115654bec","subject":"Allow db sync timeouts to be configurable","status":"MERGED","created":"2020-11-23 01:40:12.000000000","updated":"2020-11-26 11:52:01.000000000","submitted":"2020-11-26 11:50:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763703-bug/1904962","meta_rev_id":"871ee5f987590e10954633e16bb7941d33fd17c1","_number":763703,"virtual_id_number":763703,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-26 11:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-11-25 11:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-26 09:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-26 09:00:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 04:23:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 15:22:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-24 17:56:35.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2020-11-25 10:58:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-11-26 09:00:18.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"54237fb4dd99aa7131e531b203ec2180bf045a63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-23 01:40:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce0d1a76d86269376d242b698d4cecc317102694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 04:23:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb13e913b7c74651ad4c829ea7188d3b : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa343d48bfcc4123a030dea5eda3a01f : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/72e8da44605b4696a2a7a32620727150 : SUCCESS in 11m 57s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b10afcdad3514190bded9e5cabea3aab : SUCCESS in 9m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78d36c04d8ac4376a6eebead5639c71a : SUCCESS in 9m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62cd08db4c1b471bac96beae274a993f : SUCCESS in 51m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b505ee0023134ee8bfdcd15474ff9834 : SUCCESS in 1h 31m 45s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4c1bd0b102024b3dbfb3095a6cd4218d : SUCCESS in 19m 34s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60a762a479954303be8d1694f26a0aee : SUCCESS in 14m 33s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7020a2f2debd4e3f834e962e85dcc0be : SUCCESS in 2h 35m 38s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5a9c1155c73b4e51ae3715cd71b93fff : SUCCESS in 1h 25m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4279314766c24db9a4b52e81ce437a33 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a4ceba92c9e57ac331bdaf786912fada72f5036a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-24 15:22:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b80ad5da94ae991ca436afb25a38583b0d587fc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-11-24 17:56:35.000000000","message":"Patch Set 1:\n\nLooks good, but why not update ironic-inspector as well?","accounts_in_message":[],"_revision_number":1},{"id":"9f3634185db4a3a7ee7ec9471bd995046dae290b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-25 01:19:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b95278b80b4c775bb5de14b37e1551c2cc0e5df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 06:03:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6994826e9ee0477593c16d885d7fa517 : SUCCESS in 5m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4249743e4d3432c960a7b5f41d0f8e9 : SUCCESS in 5m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/56385ee21c7944fb92ca8b2dfcee13fd : SUCCESS in 14m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cfeca460038456d8ed8565c63af12a0 : SUCCESS in 17m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1eaadccecae4ec9bad4b2f6c9adee2c : SUCCESS in 12m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e25bc08735d4fcea8a0a277ab7ea920 : SUCCESS in 57m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a782e017f8984adaa09b85e43c711a85 : SUCCESS in 1h 16m 45s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4c3a1d4be04e4c4e9e1e1ee7a126e16d : SUCCESS in 20m 44s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cc4b586b4b94b50987b2794c0a96627 : SUCCESS in 17m 56s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bcc7069fba554fe993b00b8dfc5d9156 : SUCCESS in 4h 35m 23s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f09a2ec039c745359cad62f7a2537d33 : SUCCESS in 2h 54m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bba578812b07426380f7e9cc832c00d8 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"4dffaaae01245f22d5189b97f352031914bfa215","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-25 10:58:39.000000000","message":"Patch Set 2: Code-Review-1\n\nAdding -1 since my comment for another patch[1] is applied to this patch as well.\n\n[1] https://review.opendev.org/c/openstack/puppet-openstack-cookiecutter/+/763698\n\nI think we should also update ironic::inspector::db::sync , too, as Dmitry pointed out.","accounts_in_message":[],"_revision_number":2},{"id":"e363d9a762eeb14b0665d5949c69a80264466102","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-25 11:05:29.000000000","message":"Patch Set 2: Code-Review+2\n\nI noticed that we already have the db_sync_timeout in puppet-nova and neutron.\nThen we can keep the current name to be consistent with these existing parameters...\n\nAnd I was somehow looking at the old version without change for inspector...","accounts_in_message":[],"_revision_number":2},{"id":"ebd71a665129b2fe5e7f6e35ad7ee7c990239142","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-11-26 09:00:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11eabadd4567facc76c18b5f8fd7a43b11fa00b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 09:00:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"400b0da2adde74717ecf940f921d91b4949b4030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 11:50:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e807efeaa5c488390b39c0df2a29a1c : SUCCESS in 4m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44534f428b7f4ca99ae00cff1caa98e2 : SUCCESS in 3m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1ed5b5ea784c4ef689fab0fe75ce89ee : SUCCESS in 12m 50s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/686ae20b32434c609a920b929244c3b2 : SUCCESS in 9m 47s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52c68dcca46242868c91d7fe11e3ec9a : SUCCESS in 59m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a949d7a62155446d8837eb08998466fd : SUCCESS in 1h 39m 36s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/aab94eb9e7c24b95b4d5695b75e76773 : SUCCESS in 17m 47s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d64b5705b004c3e8f46ce7878f05843 : SUCCESS in 13m 58s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ea17095c8bd34485939e20d932d45b94 : SUCCESS in 2h 43m 43s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/92cb2d8b0fe943bebbde2c37e11a002b : SUCCESS in 1h 27m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9817fc8cf01a4377a62d9a484fc6e29b : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d6a2d94991061a39374ed6f09bce365c1e1708a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 11:50:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"871ee5f987590e10954633e16bb7941d33fd17c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 11:52:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e844255746374b5eb9295e4bf2ba0743 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"60eb6d732dc62e2e753626da0eed955aeaf79099","revisions":{"7f017c2f00f51866b839c9d7d20bc6a2c8257c0d":{"kind":"REWORK","_number":1,"created":"2020-11-23 01:40:12.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/03/763703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/03/763703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/763703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/763703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/763703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/03/763703/1"}}},"commit":{"parents":[{"commit":"8d64eeea0bef054fc0740514e590d8885a508d4a","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/8d64eeea0bef054fc0740514e590d8885a508d4a"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-23 01:27:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-23 01:27:52.000000000","tz":480},"subject":"Allow db sync timeouts to be configurable","message":"Allow db sync timeouts to be configurable\n\nAs Openstack projects continue to have longer database migration\nchains, the Puppet default timeout of 300 seconds for an execution\nis becoming too short a duration on some hardware, leading to timeouts.\nAs projects continue to add more migration scripts without pruning\nthe base, timeouts will continue to become more frequent unless\nthis time can be expanded.\n\nChange-Id: I7c461005b2686e50dc8affd9bf2ecb0115654bec\nCloses-Bug: #1904962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/7f017c2f00f51866b839c9d7d20bc6a2c8257c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/7f017c2f00f51866b839c9d7d20bc6a2c8257c0d"}]},"branch":"refs/heads/master"},"60eb6d732dc62e2e753626da0eed955aeaf79099":{"kind":"REWORK","_number":2,"created":"2020-11-25 01:19:24.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/03/763703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/03/763703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/763703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/763703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/763703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/03/763703/2"}}},"commit":{"parents":[{"commit":"8d64eeea0bef054fc0740514e590d8885a508d4a","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/8d64eeea0bef054fc0740514e590d8885a508d4a"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-23 01:27:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-25 01:19:08.000000000","tz":480},"subject":"Allow db sync timeouts to be configurable","message":"Allow db sync timeouts to be configurable\n\nAs Openstack projects continue to have longer database migration\nchains, the Puppet default timeout of 300 seconds for an execution\nis becoming too short a duration on some hardware, leading to timeouts.\nAs projects continue to add more migration scripts without pruning\nthe base, timeouts will continue to become more frequent unless\nthis time can be expanded.\n\nChange-Id: I7c461005b2686e50dc8affd9bf2ecb0115654bec\nCloses-Bug: #1904962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/60eb6d732dc62e2e753626da0eed955aeaf79099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/60eb6d732dc62e2e753626da0eed955aeaf79099"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
