)]}'
{"id":"openstack%2Fpuppet-ironic~794831","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fvictoria~If66e6478563f9968ce1c25d7739a4feefbe0f8b1","project":"openstack/puppet-ironic","branch":"stable/victoria","topic":"bug/1930403-stable/wallaby-stable/victoria","hashtags":[],"change_id":"If66e6478563f9968ce1c25d7739a4feefbe0f8b1","subject":"Set fixture paths for unit tests","status":"MERGED","created":"2021-06-05 00:35:03.000000000","updated":"2021-06-11 16:27:19.000000000","submitted":"2021-06-11 16:27:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794831-bug/1930403-stable/wallaby-stable/victoria","cherry_pick_of_change":794384,"cherry_pick_of_patch_set":1,"meta_rev_id":"d424aba0372fc5ef34960c0d00af7818beb06f35","_number":794831,"virtual_id_number":794831,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-11 16:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-11 15:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-11 15:07:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-05 01:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0640f877878b3dfd672087519114322f47928d23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-05 00:35:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"46adc5cde48d0b7327ae23af715e58d06cefbbc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 01:05:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a83575c0995d42cd9a894b87497a64c2 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1c0e50728914b4380aba8b4a518f952 : SUCCESS in 4m 27s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a83e436309d401a9f1814c511041c8b : SUCCESS in 14m 30s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9424a27c6cf4476da97401cab5498c29 : SUCCESS in 8m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c17bdc22aba347669e46f00678649e86 : SUCCESS in 11m 44s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ff20c1c227a346acb44e1d29a39742c4 : RETRY_LIMIT in 4m 30s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5b328cc9d8b4dd381b2c5ae5a907857 : SUCCESS in 21m 20s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffb47e8276a64d83a5d6dc7e6e4725a4 : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67f04a1c732fbff5a2488a76e6fc8030cd0e1a2a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-08 01:59:06.000000000","message":"Topic cherrypick-bug/1930403-stable/wallaby-ew0154imud removed","accounts_in_message":[],"_revision_number":1},{"id":"1d47e515b35a1049f7c8eb57de85eac2782d8f9c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-08 01:59:13.000000000","message":"Topic set to bug/1930403-stable/wallaby-stable/victoria","accounts_in_message":[],"_revision_number":1},{"id":"812cb5157d8c15b817c53dc41260dc25dae78af5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-11 15:07:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"208ca9b168b41308ab11674e2655f4a8512ad724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 15:55:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ace144a3077047919d329919e8efdef1 : SUCCESS in 5m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90520458549143eea325560c3a36a503 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b6f8db70d2a4b788f57e2155ce1d8e2 : SUCCESS in 15m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7ce9546195b43bab7a83320acfcf62f : SUCCESS in 12m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/929fbf507f544762b98ccbfbe492bd8d : SUCCESS in 12m 17s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/bccdffd9a96243f0b43a4a6a6e124949 : SUCCESS in 23m 01s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a290a2dda6944cb9231a7c5bddb0d4b : SUCCESS in 25m 23s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc7e663510294d789de5e7f8ebf137cf : SUCCESS in 17m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14130c2db307f89d83b09b1f17fbc0aeee526760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 15:56:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f046784f6097a1edc62c63f9f4cc38d55fb10fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 16:27:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e48d706391b44c3da249a9ffbeefb2d3 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc30edc8a38441078c845274d3a75d88 : SUCCESS in 4m 12s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/114ca90144304080bb2512c8aaea8ae0 : SUCCESS in 15m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c13031b50ee49bcbde72d55909621b1 : SUCCESS in 10m 37s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/7665f1b3923c49acb9b591e5a4db3ba2 : SUCCESS in 24m 51s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/50527d9ef7dd4a7fb9581e8498859389 : SUCCESS in 23m 54s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c197bb74566c4da6ac92548c8d104af6 : SUCCESS in 16m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d424aba0372fc5ef34960c0d00af7818beb06f35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 16:27:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07ef892ee9d292fef418d40a37d2cef5347350e6","revisions":{"07ef892ee9d292fef418d40a37d2cef5347350e6":{"kind":"REWORK","_number":1,"created":"2021-06-05 00:35:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/794831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/31/794831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/31/794831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/31/794831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/31/794831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/31/794831/1"}}},"commit":{"parents":[{"commit":"a65b5ffb84bfcee9527adbf35d85658ce18a3b21","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/a65b5ffb84bfcee9527adbf35d85658ce18a3b21"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-02 05:01:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 00:35:03.000000000","tz":0},"subject":"Set fixture paths for unit tests","message":"Set fixture paths for unit tests\n\nThis change defines manifest_dir and module_path expicitly in unit\ntests so that modules installed under fixtures directory is properly\nloaded.\n\nCloses-Bug: #1930403\nChange-Id: If66e6478563f9968ce1c25d7739a4feefbe0f8b1\n(cherry picked from commit cbca433a2c24b581fec944a9389f2c8decd3b30c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/07ef892ee9d292fef418d40a37d2cef5347350e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/07ef892ee9d292fef418d40a37d2cef5347350e6"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
