)]}'
{"id":"openstack%2Fpuppet-ironic~797683","triplet_id":"openstack%2Fpuppet-ironic~master~Ia30aff290ec24972f387612851f8f630ddc9403b","project":"openstack/puppet-ironic","branch":"master","hashtags":[],"change_id":"Ia30aff290ec24972f387612851f8f630ddc9403b","subject":"Clean up ipxe support","status":"MERGED","created":"2021-06-23 15:34:30.000000000","updated":"2021-07-02 12:30:11.000000000","submitted":"2021-07-02 12:28:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"797683","meta_rev_id":"7b11240574e7bb970d26849bdb15067be66645b3","_number":797683,"virtual_id_number":797683,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-02 12:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-07-01 13:45:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-01 13:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":2,"date":"2021-07-02 09:35:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":1,"date":"2021-07-02 09:35:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 19:56:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-23 20:36:38.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2021-06-23 20:36:38.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-06-23 23:51:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2021-06-24 14:39:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-07-01 13:45:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-01 13:45:45.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"e820355ac89f3b87426191aaf90aa6f921abaded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-23 15:34:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a98798a15741e04bcb9fde4fd402ba957600d0f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-23 17:20:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c16163df38ae81c98e5e480930b49516873bae61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:56:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/661ac60b4cd04fd595eda7da6a5497a0 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5861b590cd0d4574960d724e8ebff2df : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7ac00c9402c4da99c1c89cba8f56178 : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e7f94a0bb234e6b98603dfd66f26e26 : SUCCESS in 14m 19s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/626e53a54f1d4cc3af5dc20a1e2f9d82 : SUCCESS in 14m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/413b222a1d504755a2e4c525cd82327e : SUCCESS in 11m 07s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfb6aee067444b5bb747ed22f55fe4ae : SUCCESS in 11m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3c2e5278dc2445a9012a789965e45df : SUCCESS in 11m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03aa811e6d93484cb251495fdf26e96b : FAILURE in 47m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bcbd3b9dc71452aa8af2e202271407e : SUCCESS in 1h 38m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/83ab3a32ce4c467c89dbdc8fa6e1b03e : SUCCESS in 1h 26m 21s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/df9ee69ced8143efb4ec0e8d2c552707 : SUCCESS in 30m 33s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55b1777bd4e84e12a9087aca13d26f8b : FAILURE in 18m 02s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/30b0b0719beb4dc3b5bea6546bb19847 : SUCCESS in 24m 28s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc7c92ff94174ddead826527338fe44e : FAILURE in 20m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ff8d3b3c0fd41c99fc97f7ba182682c : FAILURE in 17m 43s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0d9a1d327dc74a4f9de7aed83b6ed8d1 : SUCCESS in 2h 24m 06s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2d310097b97a49ceb9b5c5aea4e730e0 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":2},{"id":"26300063db1ad3c3373aadabbd530b25c705ed51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-23 20:36:14.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0f1ecc119b2c057c665d6b0287836a6c6e82cd5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:42:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/139bef26c09245e997c50e7f2a2d32c1 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f53c20d3c5d548c7994a3400a1d80524 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b0942750e0044ab862c6e05f4a5382e : SUCCESS in 4m 12s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb77506211654d87a61cd483ecb2c82f : SUCCESS in 14m 14s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/835a2b5e1f5d46c490f8c63344ba4615 : SUCCESS in 14m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ae3fcaa3a924bcbb155bff8eee63d26 : SUCCESS in 10m 57s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feb4427c45164b91b8b03a5599ea9a76 : SUCCESS in 10m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9169a89987874780800749a8a7552f91 : SUCCESS in 9m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e9d899fa3bc412e81e708a0a8d2e797 : FAILURE in 47m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0346459e9f4a4f1db400dccd32d87696 : SUCCESS in 1h 42m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/70dc4e42ebdb4edc8167ff3555786f78 : SUCCESS in 1h 45m 29s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/93c8b26f19da4997ac5674c90d55d8ae : SUCCESS in 23m 46s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af9a41c164914126b50a9e06546a88fe : FAILURE in 16m 04s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec6d28bbe1c146a4b290c4d151e98498 : SUCCESS in 29m 24s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2b85b3c21f9446fa050bb2e5a54ea86 : FAILURE in 16m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf07b22579ea46dd8df836f8911a99b4 : FAILURE in 16m 39s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/08f476cbaa4b4de6905289bf2812c9bf : SUCCESS in 3h 00m 14s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/27915c7d4ecc4749af33c55a40c926d0 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":3},{"id":"42799dc1b097501eff04b5d33507394743577cf7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-23 23:51:39.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ee27f21a1e19a2ddce8485065d34250b7a421a05","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-24 14:39:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nAlso, could you please add a release note to explain the change in parameters/defaults?","accounts_in_message":[],"_revision_number":3},{"id":"672fb5685a7b68d87cb948f8815d0b0ba797c15a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-24 15:38:00.000000000","message":"Patch Set 3: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ca7ee15df1e3c02c4a1fadb232b8aaf5da03c2cf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-24 15:53:24.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"eee6203db5fbdb4ec72eda94b38795ce4d2d5e48","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-24 15:56:36.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"520a1e30d1577a13063f17971c11fd0e6c8f817d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-24 15:59:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7569619d1f6429b15f37614cd1e334013ea6494","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-24 16:44:01.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e099dcfaa6c214b9e4dd86d0a5b1fcc2585da7e9","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-24 17:04:32.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"808cb262faaf80f79d93b45e3a0ce635dd2f30b7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-24 17:13:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5623eeb2fad88901035028819a73b600e055f9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-24 17:26:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"52608f0a10b0615c86ddca79f9179835a474c00d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 20:07:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc0ce9381ba8459c9adc04a6382166a5 : FAILURE in 5m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b61315b19ff416f91459e747b97a8b5 : FAILURE in 6m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a23dd0cc833048ee90be2714ac398733 : FAILURE in 5m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e391f10cbe554736839e6b21fd72fbe6 : FAILURE in 18m 04s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/500059a25deb45439c97d7bfb4dbb72f : FAILURE in 18m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e07cb4f60fb41339489f3234918bf90 : FAILURE in 11m 36s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc87ae6a8a554d7bb40a921447125c4b : FAILURE in 12m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f96f77e0078341cb84dfe89f314a4cb4 : FAILURE in 13m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28476dbfb58546ce967ebb0e13627ed0 : FAILURE in 46m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9ef60c5b5164d93b11313531e8d6125 : TIMED_OUT in 2h 04m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/33029c5c6bd24aeeb727d92386e0ede0 : SUCCESS in 1h 54m 06s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7683cf7c312c4cbda7e64872d7921162 : SUCCESS in 26m 23s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24addd4c442045c3b107503cfca2c303 : FAILURE in 21m 56s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/8045be566e774783a6e82e83ff06f42e : SUCCESS in 26m 41s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb1ca8ac7f7647fc94d2b888e3f5e629 : FAILURE in 18m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/742d153a40a841d7bb2f494d2be6ff9c : FAILURE in 18m 45s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/17a5d34b12664415b00aba73382cc56b : SUCCESS in 2h 02m 42s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c7cba8b77c574d1ba1086998e22b9204 : FAILURE in 39m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e33f003982948499d317231650e2078 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":4},{"id":"81840db2319b8ed4597b802299b77e93ef4e0cb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-29 21:56:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"568dfd98460427d377ffb2c0e6fe042195e80a60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-29 22:31:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e6e0c5cef9baeca9e2276420032c87f3704633ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 00:39:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10b5c4573ca84004a12a129ebfb81781 : SUCCESS in 5m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e15daad7953544338af9f3db3264f0f1 : SUCCESS in 5m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f547087b386a49b5a31f3f68e6c5ea16 : SUCCESS in 5m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ebaac2feddb4458b51e6e6b0b6bc678 : FAILURE in 12m 50s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/40c676e95b814c149b1bc969cf55a64a : FAILURE in 16m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb37767bbabc427e8fee2a225532a7b4 : FAILURE in 11m 33s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53f6448e22084533b954bec1658fcc0a : FAILURE in 14m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5728c9a8892a495c8be447403b2d9243 : FAILURE in 11m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05b2bf069ce24dea86d0ea6f676f9672 : FAILURE in 46m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d6f937be7ba410499159ef65d7a0be6 : SUCCESS in 1h 15m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6ea1b4c12ef4480866ca3215d2dfdf2 : SUCCESS in 1h 31m 21s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f990e0a729834c91b12699c8e85d0e0b : SUCCESS in 25m 27s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/426552e1ae2846a0bb2ac61b00bfed2c : FAILURE in 20m 01s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ea7b65f3add4d84ab696f94c44f6fd8 : SUCCESS in 25m 56s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c3b0dab69df4bec85168c94bbba8b74 : FAILURE in 19m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd435f8fff85496a988065835357f467 : FAILURE in 18m 01s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/aaba7bf591f24edbb488c46c20a14d17 : SUCCESS in 2h 03m 15s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e0fcaedabe314b528b9bba8189fdbbbc : SUCCESS in 53m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95fbfc9f4c4e4abe893e1ffb94d2557a : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":6},{"id":"53021a2bc0d1d9a5ec150e10512787e04595f176","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-30 17:06:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a746ebd65928909b5206782b748226d9b11ecac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-30 17:53:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5d0ec01f1fe726eb3624be72aaf84468505e75f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-30 18:01:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7fb48f76d3d97efaf68ba463a43db50f20bf5179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 19:50:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81111eb4d4854dd29aa56041611fd0cc : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8c651eaba994548bdcb6da329743c96 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3d77cf3e5a44ddeb8f8575c4a13590f : SUCCESS in 4m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e10aba11944245dea6dfdebdc0a5754e : SUCCESS in 12m 15s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/32189387115e4b49bd23bf0fa044d4c0 : SUCCESS in 16m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6abb455bad14770ad4207cb43faffb7 : SUCCESS in 11m 20s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1e0bb86882f40c5a748ccc71145bedc : SUCCESS in 12m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c235a117cdb4476ab1ed794fa6b4b62c : SUCCESS in 11m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/192a293ee5b44828ae6c328194f0e4cb : FAILURE in 52m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8df79f99f48042e5868ddd8bc2f764e9 : SUCCESS in 1h 18m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ac6af6c978f4e28b3cd1c41462b81ea : SUCCESS in 1h 43m 29s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9a4096e73944acabe8bad4801665db8 : SUCCESS in 26m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6141e916f26f48a2bd46a0d1b33f3bf7 : FAILURE in 17m 15s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/b17aac9ea0e44690a59777449c6bb708 : FAILURE in 37m 25s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2af42f867ed4a2cb9f8f8c6ae797d29 : FAILURE in 16m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/962c5f8e9ebf4244aa6d811ddb12db40 : FAILURE in 17m 10s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/151e333cfb324774a696b3d17fb43221 : FAILURE in 30m 43s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a5ebca6e2194c4280a8ad74bc51c44c : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":9},{"id":"371904cdb9a234d65bd6e297403fc280be483803","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-30 23:11:53.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e9a3830aeee623f23ffaf690f232565a51f831fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 01:38:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8febcf5a55bc4893ba171889ca288f65 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de72c14faaad4ea5a32bf381229e3b94 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d4f366f93694c11a29d911fb879fa74 : SUCCESS in 5m 04s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d7b32528fee4be6bb88d46ce998a9c4 : SUCCESS in 16m 15s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7d15ebb686c44e793cee0a56540a2b8 : SUCCESS in 15m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80d6ef62abd14f1e93d61573fb466b0f : SUCCESS in 12m 02s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3a229d248b44e12b5c53943869feb8e : SUCCESS in 12m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18806e6e5f2b41dd9ead7f99727b99f4 : SUCCESS in 10m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70b0ce3c1a3644c9b752afb10f73445f : FAILURE in 38m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/620d320ea4a6476ab9a8168c645666d7 : SUCCESS in 1h 58m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/478c7c81defe4b3fa7a3fe772d6bcd29 : SUCCESS in 1h 36m 58s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee81bcffbed246f690f24e47ca2063c8 : SUCCESS in 23m 41s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac60c1644bf04548a5110c5ccac7a852 : FAILURE in 17m 27s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/6478eeb7d5a14833868b05159b9a2e42 : SUCCESS in 25m 07s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/577a2479527e42b4846d704d5bb076a5 : FAILURE in 17m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30bca1dbbeff4c26a212dcf89d3c0a0f : FAILURE in 23m 52s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2384b0e81187477393a45a0a5cb7a68c : SUCCESS in 2h 21m 35s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/56942eb40bd14b25864e470918cddbf1 : SUCCESS in 58m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e19b46fe1c84fbba022da98dd915620 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":9},{"id":"1cf3ef246627da3f93149029bc66c9eb1a7f05a7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-01 13:45:05.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"9b000e4ed1f5777f4d32a671c243156674ae1e8c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-07-01 13:45:45.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"69893cf927cedc1e41e507ff9b7dcfbf0dfc488d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-01 14:01:45.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a710c7c17804017dd550b309143960342f4d6626","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-07-02 09:35:02.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"8063cd36b8830348ba4c9776d6a9f23f2a5521ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 09:35:34.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"fb58a396c802e63e0aab8ea3b6514e7bf4fc86b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 12:28:52.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40e5f2ebb79e4ba085c59bdb1fd5c04e : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a4fc903001849ff8692813c61eaba16 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86c038e07a4548ee958f20b597ac012f : SUCCESS in 6m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3be10dbaeb4c40f791728eee20d47ae7 : SUCCESS in 16m 25s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/33b57c1de5ff41bc902bccdd9267ce88 : SUCCESS in 14m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c66b7764a8614c9c86c7df04b0f22e7c : SUCCESS in 13m 08s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/960f3321dfec4234bdc04bd873b1cd84 : SUCCESS in 11m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/38b4dcc9ae034b97a01852127a6e98f0 : SUCCESS in 1h 59m 33s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/afec5e554032491b98a4452d48feb5d0 : SUCCESS in 26m 53s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2135d27e29be479d901d004effd4e857 : FAILURE in 16m 35s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2c47e280b1a74f058a95cec99c81d3c6 : SUCCESS in 2h 47m 40s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e6f4a8c5185b4d189a7dabe531e6c915 : SUCCESS in 1h 06m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2f0229bffcb40cdb419e932cac79b4c : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":9},{"id":"ebfecd9fc8f72147ceaf143f846ecafe0dfeecab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 12:28:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"7b11240574e7bb970d26849bdb15067be66645b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 12:30:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0465e42035b487bb6bc4a4a90a213d7 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3864e15998b5b1eec7d2b1b4911add9bb899fdb8","revisions":{"86059e5407272ba9b7561378d76768c81f76b4f6":{"kind":"REWORK","_number":1,"created":"2021-06-23 15:34:30.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/83/797683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/797683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/1"}}},"commit":{"parents":[{"commit":"12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"subject":"WIP: Clean up ipxe support","message":"WIP: Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/86059e5407272ba9b7561378d76768c81f76b4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/86059e5407272ba9b7561378d76768c81f76b4f6"}]},"branch":"refs/heads/master"},"f764e56f31e99ca3f5d776877bc1eedf179ea506":{"kind":"REWORK","_number":2,"created":"2021-06-23 17:20:55.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/83/797683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/797683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/2"}}},"commit":{"parents":[{"commit":"12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 17:20:50.000000000","tz":-420},"subject":"WIP: Clean up ipxe support","message":"WIP: Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f764e56f31e99ca3f5d776877bc1eedf179ea506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f764e56f31e99ca3f5d776877bc1eedf179ea506"}]},"branch":"refs/heads/master"},"222c137a714584f9085f33bcabefebd88d0986a7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-06-23 20:36:14.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/83/797683/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/797683/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/3"}}},"commit":{"parents":[{"commit":"12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 20:36:06.000000000","tz":0},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/222c137a714584f9085f33bcabefebd88d0986a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/222c137a714584f9085f33bcabefebd88d0986a7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b18eb9a661cee82a694505804d333bb4e5c44dda":{"kind":"REWORK","_number":4,"created":"2021-06-24 17:26:22.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/83/797683/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/797683/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/4"}}},"commit":{"parents":[{"commit":"12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-24 17:26:16.000000000","tz":-420},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/b18eb9a661cee82a694505804d333bb4e5c44dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/b18eb9a661cee82a694505804d333bb4e5c44dda"}]},"branch":"refs/heads/master"},"85130fcc74d68b58729b7c0462884b165f7d2747":{"kind":"REWORK","_number":5,"created":"2021-06-29 21:56:09.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/83/797683/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/797683/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/5"}}},"commit":{"parents":[{"commit":"12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-29 21:56:05.000000000","tz":-420},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/85130fcc74d68b58729b7c0462884b165f7d2747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/85130fcc74d68b58729b7c0462884b165f7d2747"}]},"branch":"refs/heads/master"},"fe664526426bd61dcb9a812ae042cb1f46c075c4":{"kind":"REWORK","_number":6,"created":"2021-06-29 22:31:29.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/83/797683/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/797683/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/6"}}},"commit":{"parents":[{"commit":"12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-29 22:31:25.000000000","tz":-420},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/fe664526426bd61dcb9a812ae042cb1f46c075c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/fe664526426bd61dcb9a812ae042cb1f46c075c4"}]},"branch":"refs/heads/master"},"615cdab6225e30bb2abc7331597d7630060c939f":{"kind":"REWORK","_number":7,"created":"2021-06-30 17:06:07.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/83/797683/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/797683/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/7"}}},"commit":{"parents":[{"commit":"12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-30 17:06:03.000000000","tz":-420},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/615cdab6225e30bb2abc7331597d7630060c939f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/615cdab6225e30bb2abc7331597d7630060c939f"}]},"branch":"refs/heads/master"},"3b709386e2b423e34949432e03fc961e8d3ce510":{"kind":"REWORK","_number":8,"created":"2021-06-30 17:53:44.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/83/797683/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/797683/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/8"}}},"commit":{"parents":[{"commit":"12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-30 17:52:48.000000000","tz":-420},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/3b709386e2b423e34949432e03fc961e8d3ce510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/3b709386e2b423e34949432e03fc961e8d3ce510"}]},"branch":"refs/heads/master"},"3864e15998b5b1eec7d2b1b4911add9bb899fdb8":{"kind":"REWORK","_number":9,"created":"2021-06-30 18:01:17.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/83/797683/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/797683/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/797683/9"}}},"commit":{"parents":[{"commit":"12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/12bc9b1d5efc6f4c1dc365d45e84529bf3f7059d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-30 18:01:11.000000000","tz":-420},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/3864e15998b5b1eec7d2b1b4911add9bb899fdb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/3864e15998b5b1eec7d2b1b4911add9bb899fdb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
