)]}'
{"id":"openstack%2Fpuppet-ironic~801482","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fwallaby~I363ed1f1a5fe46442e394929451481e5bff55828","project":"openstack/puppet-ironic","branch":"stable/wallaby","topic":"puppet-lint-latest-stable/wallaby","hashtags":[],"change_id":"I363ed1f1a5fe46442e394929451481e5bff55828","subject":"Fix lint errors with the latest lint packages","status":"ABANDONED","created":"2021-07-21 01:18:38.000000000","updated":"2021-07-22 15:46:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":790597,"cherry_pick_of_patch_set":1,"meta_rev_id":"5c12a53b334c68f93dc890ca7ae565c11385cf90","_number":801482,"virtual_id_number":801482,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-21 04:12:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2021-07-21 01:29:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-21 04:12:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb108401c1f31166e0e9719b23bc017d2888ca9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-21 01:18:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"37348ff42aea694eae73902431dec0cc22bbe8ed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-21 01:29:23.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ee155798ea7dc3ef48cdb6dade60018c0c3779ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 04:12:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ef4fa1b58514bdfa4342e0cdbd52136 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2944e10e21fc4255a11923a3b2d49c1b : SUCCESS in 3m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9474fdf9d6b94054ac717d822688c1ee : SUCCESS in 16m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/323383e1bdb54ac98ed00d2f1b7ed65a : SUCCESS in 11m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0589f138430441798f5725a6e13ca39 : SUCCESS in 8m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e1b6435f3054921a6ef83582c15dd32 : FAILURE in 50m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/95210f34c69e4591865ac7ce53feac78 : SUCCESS in 1h 17m 01s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cecfa0e8a5f4a24803c1d3c72556fa7 : SUCCESS in 20m 19s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d283f61e06d4f65915a2aa3d75d9d66 : FAILURE in 17m 01s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/98b81a75f64d48f6a1db11a3d0b81edc : SUCCESS in 2h 34m 01s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a3dc958773414f73a39e06eccedd2015 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5c12a53b334c68f93dc890ca7ae565c11385cf90","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 15:46:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59d5314f461bf3c395d45927536d2dfaac5fad71","revisions":{"59d5314f461bf3c395d45927536d2dfaac5fad71":{"kind":"REWORK","_number":1,"created":"2021-07-21 01:18:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/801482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/82/801482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/82/801482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/82/801482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/82/801482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/82/801482/1"}}},"commit":{"parents":[{"commit":"122bd74ff17b7906a2c8757aab605bfb50460024","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/122bd74ff17b7906a2c8757aab605bfb50460024"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 09:08:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-21 01:18:38.000000000","tz":0},"subject":"Fix lint errors with the latest lint packages","message":"Fix lint errors with the latest lint packages\n\nThis change fixes the following lint errors discovered since we removed\npin of lint packages.\n\nmanifests/bifrost.pp:50:WARNING: Duplicate class parameter\ndocumentation for ironic::bifrost::revision\nmanifests/bifrost.pp:38:WARNING: Duplicate class parameter\ndocumentation for ironic::bifrost::revision\nmanifests/pxe.pp:70:WARNING: class included by absolute name (::$class)\n\nChange-Id: I363ed1f1a5fe46442e394929451481e5bff55828\n(cherry picked from commit cb788a3068f548d027eba447d97625939dedc0c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/59d5314f461bf3c395d45927536d2dfaac5fad71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/59d5314f461bf3c395d45927536d2dfaac5fad71"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
