)]}'
{"id":"openstack%2Fpuppet-ironic~801484","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fwallaby~Ia30aff290ec24972f387612851f8f630ddc9403b","project":"openstack/puppet-ironic","branch":"stable/wallaby","topic":"ipxe-fix-w","hashtags":[],"change_id":"Ia30aff290ec24972f387612851f8f630ddc9403b","subject":"Clean up ipxe support","status":"MERGED","created":"2021-07-21 01:21:02.000000000","updated":"2021-08-06 10:18:25.000000000","submitted":"2021-08-06 10:16:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"801484-ipxe-fix-w","cherry_pick_of_change":797683,"cherry_pick_of_patch_set":9,"meta_rev_id":"a81787ef39c51393b6fce6a1feb20afc64701363","_number":801484,"virtual_id_number":801484,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-06 10:16:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2021-07-27 21:34:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2021-08-06 08:18:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-07-28 01:10:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-03 08:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2021-08-06 08:00:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2021-08-06 08:18:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2021-08-06 08:00:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-21 01:21:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-07-21 03:35:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 17:06:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"state":"REVIEWER"},{"updated":"2021-07-22 17:11:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"state":"REMOVED"},{"updated":"2021-07-28 01:10:56.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2021-08-02 18:53:51.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-08-02 23:06:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-08-02 23:06:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2021-08-02 23:06:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-08-02 23:06:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"7e803f32d8bbf3d8797112e7b1e5fc084f04d226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-21 01:21:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"221f0d5baf20d49a3e6e6cc668eade6e6bb77586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 03:35:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38f325c797e94fee89e117a1f23b29d7 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56f02a72e28b48ec807505e234e3e0a3 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/783b125eb70844d4842807073943cb61 : SUCCESS in 11m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69cf26bf848b4e15808dce18bdd58ae1 : SUCCESS in 11m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17f72751cbe1465daa1b6d9c51335317 : SUCCESS in 9m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08e90e91230342c9b11c85f05e1dacfb : FAILURE in 40m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb1b4c1cfc91429da30ee253350be620 : SUCCESS in 1h 12m 28s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f07d3823b15407d8d793aa428278976 : SUCCESS in 24m 36s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a26e66b4a1aa4796993a99002fca91e1 : FAILURE in 16m 33s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/66a5ae5c5b0c47a0a18ee2210d488272 : SUCCESS in 2h 12m 38s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/496b41ee9ee648ffb493fa77cd045e00 : SUCCESS in 1h 00m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0014003531f7446e8333fde80a4cc574 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7f3c5897d3d7b3672471008b86220cda2af1a9f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 15:50:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2a3482d6d1935a51ca4ceff9b36871f474a7e5bb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 15:56:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eceaa8576d0c97f3e14c58cac6141c8bf60911bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 16:08:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6cd230678e18de95ebcfddaa0adf8154de7c387c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 17:02:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c7044536595ad91df9dfa7c521fa736d230b14c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 17:06:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae1e0eafbd9c5e9f4f2a6d6d7d1f8f2109631435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 17:09:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a0572caff7e811824a819c931940f39946fce20d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 17:10:45.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"3e632e28adf5db8dd18ba4051acf8aeecc3c091c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 17:11:32.000000000","message":"Removed reviewer David Vallee Delisle.","accounts_in_message":[],"_revision_number":7},{"id":"cccd42f8b8a337895dcaa0c59d6919bdbc54ccc1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 17:11:52.000000000","message":"Topic ipxe-fix-w-patch4 removed","accounts_in_message":[],"_revision_number":7},{"id":"c015c03d8fc8f82f87218e06a08f561d89efd3ea","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 17:11:55.000000000","message":"Topic set to ipxe-fix-w","accounts_in_message":[],"_revision_number":7},{"id":"bb02165f5679d20dd2c71cd6e0b259a3fcfbafc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 20:08:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bff98672f894e57987c2f4673959def : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b06d783c33494174993ebbea7a21cc9a : SUCCESS in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/867b3ca1997d49dbbedfcedce38d078c : SUCCESS in 14m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b37b56849855404aa2be9c1e3cef22fb : SUCCESS in 9m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d099055d65824b2583aaf438562878ef : SUCCESS in 10m 06s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89d06fee42c741639b8eddfbe7c0b953 : FAILURE in 46m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e8ba24417a848a98c8effb5e15d848b : SUCCESS in 1h 22m 09s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f639d676b2f34091a961d7e607d12bc4 : SUCCESS in 28m 04s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55bd958fb8cd4b35a58304d3912bcfcf : FAILURE in 17m 23s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a46db092e48445b1be6d58ac51f3b549 : SUCCESS in 2h 51m 16s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ca8a72efcd724845adf3b25836eff556 : SUCCESS in 1h 08m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2695a57e0668492ca6b544eb3f29a52c : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":7},{"id":"24627ee8683456e60a0c4848a86846a4b59d570d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-27 14:10:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"204ad30dc70a3cd0f42c6eef2cacfe98a6863d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 16:50:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f33ed03d6f944b3693e8a5b9ca750e18 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc10147e32294acaa31d7e2e9596b638 : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/80695ee545d444fda587ef81c9834148 : SUCCESS in 14m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ef09643e0df433aaaff49f226d3a0dc : SUCCESS in 10m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9df929ab3254e66b262aa448fb1508d : SUCCESS in 10m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc5170e6e3a8475692d0148250dd3e03 : FAILURE in 39m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/78b2862ee603457497a2d534fcd28144 : SUCCESS in 1h 29m 48s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/abbbb20fa0834d63b4d5e5822cc4d262 : SUCCESS in 26m 04s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8318c0f1d1224eabb04062489077d23b : FAILURE in 18m 28s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e4074a28e0a24cadb2493147c2a297e9 : SUCCESS in 2h 38m 23s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8650c4c78e4f44fe85fb2c93229641eb : SUCCESS in 1h 15m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b57810940f2c419c9869f077053a7a91 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":8},{"id":"09e6c88959239cb3f7556efa631ba40b76fbb848","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-27 21:34:48.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e4d59879fbad9481d036e2bfc85bf395568cf278","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-27 22:56:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c1c44df2a7a6a026504520052d7dabe9de0d6d4b","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-07-28 01:10:56.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"f3558f10fbe6ba111dab8c526f7c099a92405901","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-29 17:15:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"46d1baff426e3a33eefe67bd8a149bec6d8246c7","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-08-03 08:23:49.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"8d2e943c858c45f593ee87b3c8e2564c0181ba85","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-08-06 08:00:47.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ef68c4f30e68caa2c998331b1ebf094b56b1d906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 08:01:57.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f266251290cf04a2af0c0c013fc2ee4eb7154494","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-06 08:18:06.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"2bdfdd03ff752d567176419f4eea5f7f04649167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 10:16:54.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21a59ea358fc424e8b8bfba9940173e3 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69ff410d4e4a45f7baec23e8d2954ff8 : SUCCESS in 4m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcde02501916443580078db8a323fec6 : SUCCESS in 13m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b220d1e09f94e68aca77fbaf30a871b : SUCCESS in 9m 46s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cfe6180d3724ec39e966b597593c059 : SUCCESS in 1h 25m 45s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/7194e49fc54e40919abb35ca77474013 : SUCCESS in 21m 40s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a61e2b957514c4aa4eaa26d7a13f8fa : FAILURE in 16m 00s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fb04078318f84a03b680137eb733473d : SUCCESS in 2h 14m 10s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/aa951de94da74fdb866ce7eddd073a7f : SUCCESS in 1h 01m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82b115fd5a3a428ea632619d60aca48b : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":8},{"id":"adca34d72f8714789f0092edf731863d1aec6980","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 10:16:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"a81787ef39c51393b6fce6a1feb20afc64701363","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 10:18:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca9ed24bce094706b077789fcdab3dad : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"7cc374a063c84d56a5d971776e46a6157636f552","revisions":{"1d1f19c27b2f1e97c774b148d59325432b058dd5":{"kind":"REWORK","_number":1,"created":"2021-07-21 01:21:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/84/801484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/1"}}},"commit":{"parents":[{"commit":"c3aeb0260cba6687d8c442ede2ce4e775102237e","subject":"We shouldn\u0027t set pxe/ipxe_enabled in ironic config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/c3aeb0260cba6687d8c442ede2ce4e775102237e"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-21 01:21:02.000000000","tz":0},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n(cherry picked from commit 3864e15998b5b1eec7d2b1b4911add9bb899fdb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/1d1f19c27b2f1e97c774b148d59325432b058dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/1d1f19c27b2f1e97c774b148d59325432b058dd5"}]},"branch":"refs/heads/stable/wallaby"},"e7afd30d31947d320b13ba231f11f2cefcf7d397":{"kind":"REWORK","_number":2,"created":"2021-07-22 15:50:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/84/801484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/2"}}},"commit":{"parents":[{"commit":"fc299c77c9035f38168f587173b2d663097b4e2f","subject":"We shouldn\u0027t set pxe/ipxe_enabled in ironic config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/fc299c77c9035f38168f587173b2d663097b4e2f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 15:48:30.000000000","tz":540},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nConflicts:\n\tmanifests/pxe.pp\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n(cherry picked from commit 3864e15998b5b1eec7d2b1b4911add9bb899fdb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/e7afd30d31947d320b13ba231f11f2cefcf7d397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/e7afd30d31947d320b13ba231f11f2cefcf7d397"}]},"branch":"refs/heads/stable/wallaby"},"028a34ab78820d64d9c40fe6ae01afd88c21fccc":{"kind":"REWORK","_number":3,"created":"2021-07-22 16:08:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801484/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/84/801484/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/3"}}},"commit":{"parents":[{"commit":"fc299c77c9035f38168f587173b2d663097b4e2f","subject":"We shouldn\u0027t set pxe/ipxe_enabled in ironic config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/fc299c77c9035f38168f587173b2d663097b4e2f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 16:08:12.000000000","tz":540},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nConflicts:\n\tmanifests/pxe.pp\n\nBackport note 1:\nThis change includes the follow-up commit[1], which made name of ipxe\nboot filename configurable.\n\n[1] 918b3116634d1be812aa5e0359eff9119ffe87a6\n\nBackport note 2:\nThis change keeps default of uefi_ipxe_bootfile_name to ipxe.efi to be\naligned with the default in ironic stable/wallaby (and older).\n\nBackport note 3:\nThe ip_version parameter is not depecated by this backport because\nthe parameter is still valid.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n(cherry picked from commit 3864e15998b5b1eec7d2b1b4911add9bb899fdb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/028a34ab78820d64d9c40fe6ae01afd88c21fccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/028a34ab78820d64d9c40fe6ae01afd88c21fccc"}]},"branch":"refs/heads/stable/wallaby"},"58989c7ec0ea8860b2d6efdda942b35d5d4f2789":{"kind":"REWORK","_number":4,"created":"2021-07-22 17:02:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801484/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/84/801484/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/4"}}},"commit":{"parents":[{"commit":"fc299c77c9035f38168f587173b2d663097b4e2f","subject":"We shouldn\u0027t set pxe/ipxe_enabled in ironic config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/fc299c77c9035f38168f587173b2d663097b4e2f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 17:00:30.000000000","tz":540},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nConflicts:\n\tmanifests/pxe.pp\n\nBackport notes:\n- This change includes the following follow-up commits.\n\n    Make ipxe bootfile name configurable\n    commit 918b3116634d1be812aa5e0359eff9119ffe87a6\n\n    Fix incorrect handling/update about deprecated parameters\n    commit b307c40f353d6f8e965e9c9d52382fe7b0145043\n\n    Fix usage of ironic::pxe::common\n    commit c21b1e8db6ece50f19c09b38ead3f7d5ec84249c\n\n- This change keeps default of uefi_ipxe_bootfile_name to ipxe.efi to\n  be aligned with the default in ironic stable/wallaby (and older).\n\n- The ip_version parameter is not depecated by this backport because\n  the parameter is still valid.\n\n- The change in default values were reverted as much as possible.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n(cherry picked from commit 3864e15998b5b1eec7d2b1b4911add9bb899fdb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/58989c7ec0ea8860b2d6efdda942b35d5d4f2789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/58989c7ec0ea8860b2d6efdda942b35d5d4f2789"}]},"branch":"refs/heads/stable/wallaby"},"cd610e52e11b35a4f891e1b583e75f71aea0205e":{"kind":"REWORK","_number":5,"created":"2021-07-22 17:06:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801484/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/84/801484/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/5"}}},"commit":{"parents":[{"commit":"122bd74ff17b7906a2c8757aab605bfb50460024","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/122bd74ff17b7906a2c8757aab605bfb50460024"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-04 15:31:22.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 17:04:58.000000000","tz":540},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nConflicts:\n\tmanifests/pxe.pp\n\nBackport notes:\n- The following commits are squashed into this backport.\n\n    We shouldn\u0027t set pxe/ipxe_enabled in ironic config\n    commit 5daa9edd96eebde8b6866f362006fae46df27f6a\n\n    Make ipxe bootfile name configurable\n    commit 918b3116634d1be812aa5e0359eff9119ffe87a6\n\n    Fix incorrect handling/update about deprecated parameters\n    commit b307c40f353d6f8e965e9c9d52382fe7b0145043\n\n    Fix usage of ironic::pxe::common\n    commit c21b1e8db6ece50f19c09b38ead3f7d5ec84249c\n\n- This change keeps default of uefi_ipxe_bootfile_name to ipxe.efi to\n  be aligned with the default in ironic stable/wallaby (and older).\n\n- The ip_version parameter is not depecated by this backport because\n  the parameter is still valid.\n\n- The change in default values were reverted as much as possible.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n(cherry picked from commit 3864e15998b5b1eec7d2b1b4911add9bb899fdb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/cd610e52e11b35a4f891e1b583e75f71aea0205e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/cd610e52e11b35a4f891e1b583e75f71aea0205e"}]},"branch":"refs/heads/stable/wallaby"},"9c4d043b1474feb80d93c4ecd3066264c22604ac":{"kind":"REWORK","_number":6,"created":"2021-07-22 17:09:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801484/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/84/801484/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/6"}}},"commit":{"parents":[{"commit":"fc299c77c9035f38168f587173b2d663097b4e2f","subject":"We shouldn\u0027t set pxe/ipxe_enabled in ironic config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/fc299c77c9035f38168f587173b2d663097b4e2f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 17:09:25.000000000","tz":540},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this and also change the default\nipxe binary for EFI from ipxe.efi which is intended to contain\nfirmware, and snponly.efi which only has the EFI networking stack.\nIn the vast majority of cases, short of very broken EFI firmware,\noperators need snponly.efi.\n\nConflicts:\n\tmanifests/pxe.pp\n\nBackport notes:\n- This change includes the following follow-up commits.\n\n    Make ipxe bootfile name configurable\n    commit 918b3116634d1be812aa5e0359eff9119ffe87a6\n\n    Fix incorrect handling/update about deprecated parameters\n    commit b307c40f353d6f8e965e9c9d52382fe7b0145043\n\n    Fix usage of ironic::pxe::common\n    commit c21b1e8db6ece50f19c09b38ead3f7d5ec84249c\n\n- This change keeps default of uefi_ipxe_bootfile_name to ipxe.efi to\n  be aligned with the default in ironic stable/wallaby (and older).\n\n- The ip_version parameter is not depecated by this backport because\n  the parameter is still valid.\n\n- The change in default values were reverted as much as possible.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n(cherry picked from commit 3864e15998b5b1eec7d2b1b4911add9bb899fdb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/9c4d043b1474feb80d93c4ecd3066264c22604ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/9c4d043b1474feb80d93c4ecd3066264c22604ac"}]},"branch":"refs/heads/stable/wallaby"},"d7d0f7042e5828c2da2ae7b91444feebcd0835d8":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-07-22 17:10:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801484/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/84/801484/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/7"}}},"commit":{"parents":[{"commit":"fc299c77c9035f38168f587173b2d663097b4e2f","subject":"We shouldn\u0027t set pxe/ipxe_enabled in ironic config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/fc299c77c9035f38168f587173b2d663097b4e2f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 17:10:15.000000000","tz":540},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this.\n\nConflicts:\n\tmanifests/pxe.pp\n\nBackport notes:\n- This change includes the following follow-up commits.\n\n    Make ipxe bootfile name configurable\n    commit 918b3116634d1be812aa5e0359eff9119ffe87a6\n\n    Fix incorrect handling/update about deprecated parameters\n    commit b307c40f353d6f8e965e9c9d52382fe7b0145043\n\n    Fix usage of ironic::pxe::common\n    commit c21b1e8db6ece50f19c09b38ead3f7d5ec84249c\n\n- This change keeps default of uefi_ipxe_bootfile_name to ipxe.efi to\n  be aligned with the default in ironic stable/wallaby (and older).\n\n- The ip_version parameter is not depecated by this backport because\n  the parameter is still valid.\n\n- The change in default values were reverted as much as possible.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n(cherry picked from commit 3864e15998b5b1eec7d2b1b4911add9bb899fdb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/d7d0f7042e5828c2da2ae7b91444feebcd0835d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/d7d0f7042e5828c2da2ae7b91444feebcd0835d8"}]},"branch":"refs/heads/stable/wallaby"},"7cc374a063c84d56a5d971776e46a6157636f552":{"kind":"REWORK","_number":8,"created":"2021-07-27 14:10:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801484/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/84/801484/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/84/801484/8"}}},"commit":{"parents":[{"commit":"fc299c77c9035f38168f587173b2d663097b4e2f","subject":"We shouldn\u0027t set pxe/ipxe_enabled in ironic config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/fc299c77c9035f38168f587173b2d663097b4e2f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-23 15:32:56.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-27 14:10:16.000000000","tz":540},"subject":"Clean up ipxe support","message":"Clean up ipxe support\n\nThe puppet-ironic module\u0027s ipxe support does not align with current\nversions of ironic which delineated ipxe and pxe support quite some\ntime ago.\n\nThis is an attempt to reconcile this.\n\nConflicts:\n\tmanifests/pxe.pp\n\nBackport notes:\n- This change includes the following follow-up commits.\n\n    Make ipxe bootfile name configurable\n    commit d99a51d17b7ae254d03225dd763f56cd90465bcb\n\n    Fix incorrect handling/update about deprecated parameters\n    commit b307c40f353d6f8e965e9c9d52382fe7b0145043\n\n    Fix usage of ironic::pxe::common\n    commit c21b1e8db6ece50f19c09b38ead3f7d5ec84249c\n\n- This change keeps default of uefi_ipxe_bootfile_name to ipxe.efi to\n  be aligned with the default in ironic stable/wallaby (and older).\n\n- The ip_version parameter is not depecated by this backport because\n  the parameter is still valid.\n\n- The change in default values were reverted as much as possible.\n\nChange-Id: Ia30aff290ec24972f387612851f8f630ddc9403b\n(cherry picked from commit 3864e15998b5b1eec7d2b1b4911add9bb899fdb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/7cc374a063c84d56a5d971776e46a6157636f552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/7cc374a063c84d56a5d971776e46a6157636f552"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
