)]}'
{"id":"openstack%2Fpuppet-ironic~804403","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fwallaby~I9256025888f5d326e34f13a087c9fdee23f97598","project":"openstack/puppet-ironic","branch":"stable/wallaby","topic":"bug/1939709","hashtags":[],"change_id":"I9256025888f5d326e34f13a087c9fdee23f97598","subject":"Fix the wrong function name","status":"MERGED","created":"2021-08-12 14:57:35.000000000","updated":"2021-08-18 04:36:55.000000000","submitted":"2021-08-18 04:36:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804403-bug/1939709","meta_rev_id":"ab2999e6d03411694eeced556fafb1ecc0df5406","_number":804403,"virtual_id_number":804403,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-18 04:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-08-18 01:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-08-12 15:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-18 03:13:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-08-18 01:41:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-12 15:10:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-08-12 16:26:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 01:41:10.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2021-08-18 03:13:00.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"9ee2f87197e4df86afa416b39cf2779970a39c94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 14:57:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8bd5fe2a2dca4a47983ca3dfb830453d5a6c214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 15:00:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ba16d88cda420b00e1b046d9b07259b9e8b7cb55","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-12 15:10:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3c9e9f0e96530ab9efa097be5fad45f01ff48e66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 16:26:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/056b81c560024d1390bc5394dd7fb110 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d51dda121a444e19b82387ab3cefdca : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9efbeda1823e4212a1f2a9c652f9a00d : SUCCESS in 13m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79501afb893f446392a1ef5e00cf7041 : SUCCESS in 9m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1707783932e4609999161e83bcba50c : SUCCESS in 10m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9db87d2d00943a1a549fe2920c86819 : FAILURE in 38m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3122e05fad041c0b2ba5fd2062252c8 : SUCCESS in 1h 24m 14s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/141792d61d9d43daba78b16edc095bb1 : SUCCESS in 21m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ca8a5fa16024858b88316a844388972 : FAILURE in 18m 00s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/00c4463c680f4f068182e2358118c15a : FAILURE in 52m 38s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"3c98037029a92e32231672fe84e83296f8f37a39","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 23:01:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cc58e81b4cbb4eeb3904c2266e67967ec67ab7a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 01:23:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13dc53858eec4e4ebb061eb3bcde6541 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a1a2d04f3e644128164909d94c08232 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0dd1c57d77dc43aa9f3462da0c03c816 : SUCCESS in 12m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/325b03703eae42379f6015c18878f965 : SUCCESS in 10m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4b069de47974c179f2d4709966d1abe : SUCCESS in 10m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c5238c4e1fc491c914ddc95171cc716 : FAILURE in 39m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfcec7d4ade748548c138b58d31ae1c3 : SUCCESS in 1h 30m 22s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/ada4b82d2b9345ae9c04507aacc12275 : SUCCESS in 33m 41s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e69bf436e0a467689bfc9e5ba56f7df : FAILURE in 15m 40s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/555d6b5e92124fe790046caeec591cd2 : FAILURE in 2h 12m 37s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"71e8ae4e4af851e79ab0a6b47c8215d13d78e4a2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-17 11:06:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"644b4e5513093c81824153645662e46e2261aee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 13:43:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/990bf095942d4509b770f923007d6d1d : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7851898b60c45799fda3d8ebabeeca7 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5311e69ef3c54c0d9a1a9c927ce04b04 : SUCCESS in 13m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5400f70b0f6c4daa831303ea51081e83 : SUCCESS in 9m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/599ecbad05b54a228727448a44095ea3 : SUCCESS in 10m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8574252914fd451f93244b5cca8177f7 : FAILURE in 40m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/041b6d03fad844a397294cab3583aa8b : SUCCESS in 1h 31m 51s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/727cb324e206409e9b4a8e9347672fe5 : SUCCESS in 21m 46s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/278c992c8df94bb9953d9681613b37ea : FAILURE in 17m 05s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e14ab555bf0940e9a4ffb6d11f28a39b : SUCCESS in 2h 31m 15s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/23c665696c244720abb6266d583ce2cd : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":2},{"id":"cd177a1dbddf026f84d4ab9e7ebc6293e5eb1420","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-08-18 01:41:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"53f2aeafb0092dfc20b0553c15a707fe9b6b826e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 01:42:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"34cb36d61254e99dd5f24983162317587c16ebe1","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-18 03:13:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"10c805d9123e5ef4e8bccb61627874c91843e26b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 04:36:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/118586c42a0041da9a4dd79abf175a3b : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5c5f21689ab4c8d97025e80be56fc35 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5c5f77441ea46b89f1a693cda80302a : SUCCESS in 14m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd85204c16744927b44547b006a05cb1 : SUCCESS in 9m 49s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c35656c1bf244a781cbcee66c15452d : SUCCESS in 1h 16m 37s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/e63c91bd75414f9ba1c6f1aaf9268fc0 : SUCCESS in 20m 29s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ccf3dc7770e4075be633d0f708ce9da : FAILURE in 17m 16s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2b97a941d2a24401929761502d596572 : SUCCESS in 2h 07m 49s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1aa827b1de5a4195beaae5c7a8dd2a68 : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ab2999e6d03411694eeced556fafb1ecc0df5406","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 04:36:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57e0c4639709126325fb2bf27d1647131cff90e4","revisions":{"bb6211158bce7d33bdfbd82e105ce6a18801c40f":{"kind":"REWORK","_number":1,"created":"2021-08-12 14:57:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/804403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/03/804403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/804403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/804403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/804403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/03/804403/1"}}},"commit":{"parents":[{"commit":"a7f1c26d984717042bed04276dc479424884d486","subject":"Prepare a new Wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/a7f1c26d984717042bed04276dc479424884d486"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 14:54:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 14:57:23.000000000","tz":540},"subject":"Fix the wrong function used for warning message","message":"Fix the wrong function used for warning message\n\nwarn() doesn\u0027t exist and we should use warning().\n\nChange-Id: I9256025888f5d326e34f13a087c9fdee23f97598\n(cherry picked from commit c8e00eec9b9ab26eb54169438ad1f238727fd069)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/bb6211158bce7d33bdfbd82e105ce6a18801c40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/bb6211158bce7d33bdfbd82e105ce6a18801c40f"}]},"branch":"refs/heads/stable/wallaby"},"57e0c4639709126325fb2bf27d1647131cff90e4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-12 15:00:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/804403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/03/804403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/804403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/804403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/03/804403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/03/804403/2"}}},"commit":{"parents":[{"commit":"a7f1c26d984717042bed04276dc479424884d486","subject":"Prepare a new Wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/a7f1c26d984717042bed04276dc479424884d486"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 14:54:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 15:00:42.000000000","tz":540},"subject":"Fix the wrong function name","message":"Fix the wrong function name\n\nwarn() doesn\u0027t exist and we should use warning().\n\nCloses-Bug: #1939709\nChange-Id: I9256025888f5d326e34f13a087c9fdee23f97598\n(cherry picked from commit cc96b3afa9bd904944c7e4b2eb8a6762fdb7f93c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/57e0c4639709126325fb2bf27d1647131cff90e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/57e0c4639709126325fb2bf27d1647131cff90e4"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
