)]}'
{"id":"openstack%2Fpuppet-ironic~819656","triplet_id":"openstack%2Fpuppet-ironic~master~I16c3368e443b7f825608e74fbf6f48fa9473078a","project":"openstack/puppet-ironic","branch":"master","topic":"inspector-apache","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-11 16:52:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I16c3368e443b7f825608e74fbf6f48fa9473078a","subject":"Support running ironic-inspector by httpd+mod_wsgi","status":"ABANDONED","created":"2021-11-29 16:04:45.000000000","updated":"2022-03-11 16:52:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"48f9ca054bd6a062903b1bf63a8e5ca1698c6e1e","_number":819656,"virtual_id_number":819656,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-29 19:01:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2021-11-29 16:05:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 19:01:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13f3a36f618de836accc3c346e561503615143f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 16:04:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68e083eaae2c274721e565efaa8a5eb007249d18","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 16:05:25.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1101defb3ddde3f5acc54d481598103dd54ef7af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 19:01:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/476152f51ebc4865a4f7dcbc538230ec : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a792aecadc042e4a1ab7817ef14b288 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee1e9391baf14f04aa7f755069ba7d9d : SUCCESS in 3m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3cc0bdb565b14ee5ab996a68ea17a5d6 : FAILURE in 12m 29s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5640e8a22b2549ca8fdbbf3fb9e97f15 : FAILURE in 13m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dc488b1b4984403a3b1b560f6a74fe5 : FAILURE in 10m 31s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e59fa0f554d4ce7bccd99ad9897d275 : FAILURE in 9m 45s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72dcf0146eb345aab67093239cdf9d5f : FAILURE in 10m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/868bfe20f5e04e8894956a4298dfd41b : FAILURE in 1h 20m 52s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4044aac4587c42ddbfa7a4996ef02c14 : FAILURE in 11m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e27f77a4e83c41b980d3a85f81293821 : FAILURE in 10m 10s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6986e3e0721b4a7794117dd714fa1b17 : SUCCESS in 16m 11s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/baa769cc5d1d4d47b9c0edac7571806b : FAILURE in 10m 16s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fbb87b813df4e458b1de4ebd67feb69 : SUCCESS in 16m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/806c450fc3a44eef9bf4902cc717792d : SUCCESS in 15m 16s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/deadbc17644e4b8f86b0fdbcbf2b4642 : SUCCESS in 2h 48m 14s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6cb503545dcb4b3dac226feef1768aa5 : FAILURE in 1h 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b65c31fde994d31a2635ac0da0bb1b2 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"48f9ca054bd6a062903b1bf63a8e5ca1698c6e1e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-11 16:52:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9f78028a5a0a941086ad17d63f2352413a4f3b6","revisions":{"e9f78028a5a0a941086ad17d63f2352413a4f3b6":{"kind":"REWORK","_number":1,"created":"2021-11-29 16:04:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/819656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/56/819656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/56/819656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/56/819656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/56/819656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/56/819656/1"}}},"commit":{"parents":[{"commit":"7a6c766c85bd5507111d32157a454c933098c06f","subject":"Merge \"Avoid usage of custom os_package_type fact\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/7a6c766c85bd5507111d32157a454c933098c06f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 16:04:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 16:04:11.000000000","tz":540},"subject":"Support running ironic-inspector by httpd+mod_wsgi","message":"Support running ironic-inspector by httpd+mod_wsgi\n\nChange-Id: I16c3368e443b7f825608e74fbf6f48fa9473078a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/e9f78028a5a0a941086ad17d63f2352413a4f3b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/e9f78028a5a0a941086ad17d63f2352413a4f3b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
