)]}'
{"id":"openstack%2Fpuppet-ironic~819849","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fussuri~Iad32434790a1744249b284937c6b67e9b58a235f","project":"openstack/puppet-ironic","branch":"stable/ussuri","topic":"bug/1941908","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2021-12-02 07:29:36.000000000","reason":"Harald Jensås replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-08 15:33:16.000000000","reason":"Takashi Kajinami replied on the change"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2021-12-08 18:34:48.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2021-12-08 18:34:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad32434790a1744249b284937c6b67e9b58a235f","subject":"inspector dnsmasq: make sequential-ip configurable","status":"MERGED","created":"2021-12-02 01:07:15.000000000","updated":"2021-12-08 18:36:47.000000000","submitted":"2021-12-08 18:34:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819849-bug/1941908","cherry_pick_of_change":819848,"cherry_pick_of_patch_set":1,"meta_rev_id":"a4eb7c24b2cc7eb3e19b3a2848a40e01f5e74675","_number":819849,"virtual_id_number":819849,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 18:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2021-12-02 07:29:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2021-12-08 15:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2021-12-08 15:34:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 01:07:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2021-12-02 04:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-08 15:33:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-12-08 15:33:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"fa290218b86ddf786e2a73ea6389a29154a016e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-02 01:07:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"fa816a9636a08c44669358d1920d1bfbcc12108c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-02 01:07:49.000000000","message":"Topic bug/1941908-stable/wallaby-stable/victoria-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"8fe2a1a67658299273c97dbf696e01de53cdb3d3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-02 01:07:51.000000000","message":"Topic set to bug/1941908","accounts_in_message":[],"_revision_number":1},{"id":"9872d1a90a1a5c475ad9bdad919e6dca6cff8805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 04:40:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c4414b4618d44a38bcdd3b0e9011d34 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62f5e4e1024641318bf300e69ed7d771 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47251d53ab2246d2b3cba2fb46d4e238 : SUCCESS in 3m 41s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/57a76080f0cc4e778a2b90c571ce4ec5 : SUCCESS in 13m 24s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae28ac1c996d4670833e99d2d016a5de : SUCCESS in 14m 21s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e730775414c2407994d9b9b3cc91ec34 : SUCCESS in 11m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3d7d7530cf2487eb654eb33eba43d59 : SUCCESS in 11m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84186a0de59d4962a928fcbeac64ca3a : SUCCESS in 1h 02m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d332e8433e3447a58612525e60d534fb : SUCCESS in 1h 25m 26s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0624779c36e461c8f1429dc26590c10 : FAILURE in 22m 23s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/289379476d144bdfb8968d05118f6ea4 : FAILURE in 6m 57s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/60c71c194ddf43fb86b1addbad7b93d5 : SUCCESS in 2h 58m 28s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2eb5df52eefe4532a8164a23f2e544ea : SUCCESS in 1h 26m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9005ba9d3cb44ad1ad0dcca585142b2e : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3a582bc5640adf18444707ad95bc8a5d7318a8cd","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-12-02 07:29:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d8a45b0abea72360a16d889f9f6aa283365a612","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-08 15:34:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c4164908bb12fc4f20b1dc1d29a743fef0a35ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 15:35:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a59c304d649bba9146c02ca931f0a1d74d75a283","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 18:34:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2a6e6dec2fe4588ae3eaa00542c03d3 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1d5732b95984313b325380d3e8f8a61 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/028efb08153e49eaaab4d5a8db6ec44f : SUCCESS in 4m 00s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cc82739a1784a60b84951fdde97b576 : SUCCESS in 16m 22s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/c01937d5bd044d528857dda9e5016699 : SUCCESS in 16m 46s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98023cf68e4a42f1bdb9200ed6e2a338 : SUCCESS in 12m 11s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3bd7b2fd97648e197750be92ee0e761 : SUCCESS in 1h 10m 28s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/292b822e344e4a62ad8233441a244f62 : FAILURE in 26m 58s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b214f6b1e0a349aca539c65439a097f3 : SUCCESS in 2h 56m 40s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f7404373fa354889a4c00158f143f8ed : SUCCESS in 1h 47m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68a175c079ea40288532fe54f08820d1 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d954c11b3c05194c59cc1baad843c3695abe176a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 18:34:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4eb7c24b2cc7eb3e19b3a2848a40e01f5e74675","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 18:36:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b47e3d38388847b48473a0da3ae8fa21 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03e3968addced8cde8dd5e012b29329c90e6b014","revisions":{"03e3968addced8cde8dd5e012b29329c90e6b014":{"kind":"REWORK","_number":1,"created":"2021-12-02 01:07:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/819849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/49/819849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/49/819849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/49/819849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/49/819849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/49/819849/1"}}},"commit":{"parents":[{"commit":"4e4ffa031ef8c304d5bfbeb1030022be979109f5","subject":"Prepare the final stable/ussuri release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/4e4ffa031ef8c304d5bfbeb1030022be979109f5"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-27 16:21:49.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-02 01:07:15.000000000","tz":0},"subject":"inspector dnsmasq: make sequential-ip configurable","message":"inspector dnsmasq: make sequential-ip configurable\n\nAdd the parameter ``dnsmasq_dhcp_sequential_ip`` to the\nironic::inspector class. When true (the default) dnsmasq\nis configured to serve addresses in sequential orders.\n\nSome DHCPv6 clients don\u0027t operate well with a DHCP server\nthat serves addresses sequentially. The new parameter\nallow disabling sequential addressing to improve\ninteroperability with DHCPv6 clients in UEFI firmware.\n\nCloses-Bug: #1941908\nChange-Id: Iad32434790a1744249b284937c6b67e9b58a235f\n(cherry picked from commit 4f67b341eef9f0a84afefdbe63133c3449d2e1eb)\n(cherry picked from commit dc1ebc4ff5256454780a079f8e2c5139489221fb)\n(cherry picked from commit 09196e4ae54af5b17180c67dd565ac6455522fc7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/03e3968addced8cde8dd5e012b29329c90e6b014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/03e3968addced8cde8dd5e012b29329c90e6b014"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
