)]}'
{"id":"openstack%2Fpuppet-ironic~828306","triplet_id":"openstack%2Fpuppet-ironic~master~Id1aa5f3df8dd565c81ff4fda0f9bbd07faa9c7b5","project":"openstack/puppet-ironic","branch":"master","topic":"amqp_allow_insecure_clients","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-10 04:02:31.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Id1aa5f3df8dd565c81ff4fda0f9bbd07faa9c7b5","subject":"Remove deprecated amqp_allow_insecure_clients","status":"MERGED","created":"2022-02-08 13:09:05.000000000","updated":"2022-02-10 10:10:04.000000000","submitted":"2022-02-10 10:08:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828306-amqp_allow_insecure_clients","meta_rev_id":"134d29b74d6dce29823fcd519a311ca47f2f63c2","_number":828306,"virtual_id_number":828306,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-10 10:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-09 20:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-09 08:47:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2022-02-09 20:13:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 16:02:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 08:47:49.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-02-09 20:13:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"698e2d40829115aa6b7ee77191c75f3d6bb7f7cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 13:09:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4f9894859f99f67dd69cb4b5294be2414849b47","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 13:13:48.000000000","message":"Topic cleanup-db-params removed","accounts_in_message":[],"_revision_number":1},{"id":"fd4fe7cfd498b24a05ed6d55e48d80073301e2d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 13:13:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5484052c3ba6ce23a6b172e36e7de21f395417dc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 13:21:10.000000000","message":"Topic set to amqp_allow_insecure_clients","accounts_in_message":[],"_revision_number":2},{"id":"ce8b86ddfe725f5bca3900b3f354a1ea982437af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 16:02:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e8bc018f78c493ca7741c0640e9a3c7 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02471f90771446df92bf5c0a4f9e6d20 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b4028957c704f6691d50d9a6ed81df5 : SUCCESS in 4m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c9876ebc4494c84863692b4fe7fcc92 : SUCCESS in 17m 06s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b2877fb455b476c9939dde1165c35c9 : SUCCESS in 15m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9026420c60642f2a60e502ffb2defee : SUCCESS in 12m 42s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85e129dd31e74cca82f5aa14a51bdf4f : SUCCESS in 12m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b340b98d46140128e24abaab0f27509 : SUCCESS in 12m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7b7fe63f9b445b7b7f509d4783f9075 : FAILURE in 49m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/60e8d90d499d41d0bec869c1c0f34bac : SUCCESS in 2h 01m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0835d80418e04b6cadaddec0209832b5 : SUCCESS in 59m 13s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6ef546066164f8fbb5c22435809371b : SUCCESS in 23m 21s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/780d815b6fa24c96af49c9009deaf8fc : SUCCESS in 23m 57s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/989b99923f384ae1b9a5fda1efb68c35 : FAILURE in 12m 43s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00058179bf244d64abf3b94a7adfa376 : SUCCESS in 18m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/038bd1e0de28420dac278049806dffe1 : SUCCESS in 18m 00s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5e210070c2744255ac1b0619682a70f5 : SUCCESS in 2h 25m 09s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e7e0056c72cf46ae8f03f26b2f217b77 : SUCCESS in 52m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66f5d10ac9ff4c098e31c04a0a6f5660 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"0e63f91a495b82df7ce9fa68c63a8bb1c191606c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-09 08:47:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62fa8265b43494181210968650b1514cbaaf6715","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-09 20:13:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eeae667dc1564c25ad448df415e965600ace98b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:15:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c24e330fd45b51b1b8e4e88b6c4e3b24fd4aa0d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 22:29:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08b8549c6839470184e5e1f9fbd94b96 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed8511f3c2a14f0581365c67a21660f8 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73953fb3bf7b416194872db9a25deb18 : SUCCESS in 4m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/392e07693a5c4f4dbcac2289d601e166 : SUCCESS in 16m 46s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f213dceb1f6490c9d0a077818fb0586 : SUCCESS in 15m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ea575f67dfd4f31a6757111e99aa101 : SUCCESS in 11m 42s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/720c4bf1e7004de58c5446e3b7e9be99 : SUCCESS in 13m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e87f8c36dda344ba854bda277430da24 : TIMED_OUT in 2h 06m 37s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0609d2a6823448d2a78546aa34ce675a : SUCCESS in 25m 56s\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4afc21083eee4cddbc6d7724ff5bd652 : SUCCESS in 1h 51m 32s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d26230c5fa9b43ba9d95d65a642229c1 : SUCCESS in 40m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/778a60b7adf441e6a0d60e39de1298ff : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"89a591afede442223f57c92edcb5f27abe0c481d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-09 23:03:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06e04e94066391d20d2df74f091662eb8a96752d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 01:16:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4783a757d76740d9b6f928c00fb096c8 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a83c4f20cfa407f956e33aabcb95fe5 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3b9dffe1ec04d5884c93ae6403ff0fc : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/db3fb3867c1c416ead7331fd4eb2063f : SUCCESS in 15m 57s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3639d908611e4f9e9abd1fa18a5e762e : SUCCESS in 16m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34a6dbbf2d274e8782810a8af291a335 : SUCCESS in 14m 02s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/818892492b1a4a5d999c6a16be2219fd : SUCCESS in 13m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d89b3e482197410da6a20a0075334576 : SUCCESS in 12m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d088d678143442bb28e99a7d2d5ecf4 : FAILURE in 41m 20s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7b901ce4dc64621b49133b87192941f : SUCCESS in 1h 42m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc16bd61a0424deaa06f154f1a62e39e : SUCCESS in 1h 16m 30s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d47d12708c4a04b138ef8c9f268d1f : SUCCESS in 20m 02s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b47235bf64f4fbdb69fe2fe4f51165e : SUCCESS in 26m 13s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8c2b95e6eab4bc48bb4adf508b7d05a : FAILURE in 14m 35s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3dc34a0c66c476cadee7c7dc805f679 : SUCCESS in 19m 58s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68ece2a96c5d420d9765a8fdc5c67a49 : SUCCESS in 18m 14s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6151378854f84befad468f15bb84115d : SUCCESS in 2h 08m 52s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3a823222697245a2a7a5890496eca3da : SUCCESS in 45m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60432b9b5545409083afcb80148cbf9e : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"68c31ad42ae2b5013791328e3f04a373ad5474ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 01:18:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc2385d9f3972f1bc9c08f4be800181ef8edfcf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 03:52:48.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddbd464db4d3407db9f73c7ecc806293 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb41aeaddaf84aaf8726e4504cc9c592 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/562927df3cf3454b9e64602a9ba3e330 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3046fa34bcb4ca4b437830a5eb9d34c : SUCCESS in 13m 23s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc8d4c9f96c24e2daf3aead7dfcd9cb8 : SUCCESS in 13m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ef8a8a008bf4081b4e9f01615128529 : SUCCESS in 12m 01s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/123541c53b474775baf15881e19d4b74 : SUCCESS in 12m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a3e7610f0f14f6291b57efdb476f224 : FAILURE in 1h 48m 24s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/78b8e2eae0a340d6812c8cfa3692785f : SUCCESS in 21m 35s\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/84bb2cfeffc24299a55afc06687147e5 : SUCCESS in 2h 30m 01s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/862501db92484479bae442331324cad3 : SUCCESS in 54m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cabf948dd3b4a289f4b7f1f1c2fec01 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"87d94a71395e921559d209652dcee0f6c4821f02","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 04:02:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9af92b230dbaddd9d85977709f1e9f8c44ae3109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 07:20:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad17d4d4a51473fadc5a901ee414bd7 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed2bef2b94f4671a91c7a94de815b60 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/191a7659f45348a69f2f7ea8c1bde039 : SUCCESS in 6m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e51600975df24b79ba8ab5308ad20387 : SUCCESS in 13m 03s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9284094477a4209922a3323dc7fd69e : SUCCESS in 14m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8376bdb0d8a439e83a791c4c0f87bb3 : SUCCESS in 15m 57s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74ffec83e31645e88e6473809df272e4 : SUCCESS in 16m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c5f8abcf0124c8d9a41b6a9562a184a : SUCCESS in 10m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d9aaa25dcc54d48bdaef4e673723fac : FAILURE in 39m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d635bb37b1d14a288b4c318422e4e600 : SUCCESS in 1h 49m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ece979ed71e94b5e9e4739c24858b97d : SUCCESS in 59m 16s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e08051ab2f57433aa893af2058ed2f62 : SUCCESS in 15m 51s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1768396eb5994db0912475a5d46e30e5 : SUCCESS in 24m 08s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/375257e5d4f0462696050de2f1031c8c : FAILURE in 9m 42s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b75bec4bdbc14e7fb1217c8ba3c001e6 : FAILURE in 24m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/329ef0cfcfef4808bcfbd88ef9ee2f48 : SUCCESS in 18m 54s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6c1ae17a5ea3408dbefac0cb0a004829 : SUCCESS in 2h 31m 01s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9d84977fe4d34bb59a3f0396a4deaf70 : SUCCESS in 53m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5716cf3c2b84302aeb8897f1d317420 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6493d63ab30bbfe369e51ef4c5f4a9923408a6a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 07:22:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7f9d03cdd82bc42a5328b58ef0fbba517ba8c359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:08:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86ff493d3fd74d8e84f9cbfea353491e : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0dc73b61abf49b2ae79a45e9fdd7631 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9794b94815214860a0bac4a86bc0a456 : SUCCESS in 3m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/49db611d98bd40ff83e594be1c54cd80 : SUCCESS in 11m 21s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/41f6cf93d29044a78b286e98fa176304 : SUCCESS in 11m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcfd413f10a748ad8d8936f680af925b : SUCCESS in 10m 20s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff12d3ae852848a3b6456d47b057fa8a : SUCCESS in 9m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/97d5c24ca275468f92f1f8182a08c168 : SUCCESS in 1h 41m 18s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/30265fb759434b7693644db4d5d68d2c : SUCCESS in 22m 15s\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b6bafea8c94740caa5816e59d0581bfd : SUCCESS in 2h 42m 17s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/238d144059c249bc8ac25863af1f4820 : SUCCESS in 1h 03m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5195886537634f4ba177cf5f7dc8effd : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"068d05a4fe6d263257f5b816b687ca12dbfc99dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:08:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"134d29b74d6dce29823fcd519a311ca47f2f63c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:10:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fd41cb41f18410f9e4e50496367de52 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"41b1408e48accdd1356059fff88f424a118de56a","revisions":{"18949bfa05230196bd1b54b4bf7856e6f3274ddc":{"kind":"REWORK","_number":1,"created":"2022-02-08 13:09:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/828306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/06/828306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/06/828306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/06/828306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/06/828306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/06/828306/1"}}},"commit":{"parents":[{"commit":"f27dd22cbdc4876f35d0db557adb3b376d9abd5a","subject":"Clean up deprecated database parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f27dd22cbdc4876f35d0db557adb3b376d9abd5a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:08:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:08:52.000000000","tz":540},"subject":"Remove deprecated amqp_allow_insecure_clients","message":"Remove deprecated amqp_allow_insecure_clients\n\n... because it was deprecated during Wallaby cycle.\n\nChange-Id: Id1aa5f3df8dd565c81ff4fda0f9bbd07faa9c7b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/18949bfa05230196bd1b54b4bf7856e6f3274ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/18949bfa05230196bd1b54b4bf7856e6f3274ddc"}]},"branch":"refs/heads/master"},"41b1408e48accdd1356059fff88f424a118de56a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-08 13:13:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/828306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/06/828306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/06/828306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/06/828306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/06/828306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/06/828306/2"}}},"commit":{"parents":[{"commit":"cd9bf2b47311586006dcba8921d9f002f5417030","subject":"Merge \"Clean up deprecated database parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/cd9bf2b47311586006dcba8921d9f002f5417030"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:08:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:13:51.000000000","tz":0},"subject":"Remove deprecated amqp_allow_insecure_clients","message":"Remove deprecated amqp_allow_insecure_clients\n\n... because it was deprecated during Wallaby cycle.\n\nChange-Id: Id1aa5f3df8dd565c81ff4fda0f9bbd07faa9c7b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/41b1408e48accdd1356059fff88f424a118de56a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/41b1408e48accdd1356059fff88f424a118de56a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
