)]}'
{"id":"openstack%2Fpuppet-ironic~848201","triplet_id":"openstack%2Fpuppet-ironic~master~I11274e444c230c62210e2ed5c0f4811f019b16ae","project":"openstack/puppet-ironic","branch":"master","topic":"apache-headers","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-06 16:08:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I11274e444c230c62210e2ed5c0f4811f019b16ae","subject":"Expose headers option of apache::vhost","status":"MERGED","created":"2022-06-30 01:30:25.000000000","updated":"2022-07-06 16:10:15.000000000","submitted":"2022-07-06 16:08:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848201-apache-headers","meta_rev_id":"a2f81e0b5979bffd29f80271bbd6ce1df5224bec","_number":848201,"virtual_id_number":848201,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-06 16:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-07-06 14:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-05 10:51:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2022-07-06 14:20:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-30 04:25:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-05 10:51:31.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-07-06 14:20:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"cf3386da7569d6335ebd66b1d2f7dbd3cbfb519c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-30 01:30:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"870bbe99efe37dc9de0a328900090aebb8e613c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-30 01:31:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca1bf8441c3c21cdc2103b49697cf84ff5b340ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-30 02:03:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"135fcc36582b5b2531127b1382f859f94d5c9433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 04:25:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a032c780103b4873b0d6886215058d5f : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e3d5543852f4fa6b063c1fbed31ee7d : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c4956e96b184e9987075a3770a09dc4 : SUCCESS in 3m 31s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/976b2270af8a4253bc3c0966b72d3d37 : SUCCESS in 13m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec2d59d6018c484da973e74f132b008a : SUCCESS in 11m 14s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/086aef9e4dd244ce823982af91093f25 : SUCCESS in 9m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0689e97ce0e54f828120367dce763862 : SUCCESS in 9m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd3180c9c4f94595954fa00d1fbd5b91 : FAILURE in 36m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99b1c76c8cd1424d928acc4e6828fced : SUCCESS in 1h 09m 12s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5abd1d2664c4845acb757156a904231 : SUCCESS in 17m 54s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7826bbd6f1a4c1d8f82e0a9b5b9f74a : SUCCESS in 22m 06s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa5c8acb96b9493084df4eb75d6209ac : SUCCESS in 17m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09321a1ed2ab429589c5e62186c8fa04 : SUCCESS in 21m 45s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0e22c2b46dc74d1981f96d970d84fbc4 : SUCCESS in 1h 57m 43s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/53cfaa6a0bc949178d263ad46da39df0 : SUCCESS in 55m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9b3fc4b6e404004b941260536e779e7 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"6847a7469af3aebfb5f0608939b872c1ec8c7244","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-07-05 10:51:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"06d0da140c61111838ef5a522a978ca7554e5405","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-06 14:20:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b8ce1d4eda277561be63f604a4aff19b63af69b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 14:21:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5328fb6f04ac682febbcc4baed94a3043326ea03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 16:08:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b52714b19214e8eb8cb447c265eae06\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7459adbb4d1941538cedafe20c1edfe2 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7d2790f210a4a57b78931266860b51d : SUCCESS in 3m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a96ef0f66a88446fbc482e366e820e54 : SUCCESS in 3m 10s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c7bf7865927440bbc9f91605e786c6b : SUCCESS in 12m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f92cc5996e20424594d163d420a93195 : SUCCESS in 9m 35s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/025bb09ac5d6478698bad7bc5f07a75f : SUCCESS in 8m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7d98a84c4e045a18f579ba3026723fb : SUCCESS in 1h 07m 27s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/eda038b02a3c42d7845176abcb44cc49 : SUCCESS in 20m 29s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/28224bbaa0f84554ab100d740f1a442d : SUCCESS in 1h 45m 47s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bc1857c12d674cee9e83166d059e4491 : SUCCESS in 43m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cefd2995bba4e2a87501ba24489892e : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c3cbb092b681f544c4ca069cdc597fdc222df84e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 16:08:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a2f81e0b5979bffd29f80271bbd6ce1df5224bec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 16:10:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a1d086c7e4463fb0204098cdd55bda\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4de9022c6fe64ae6a55fa1c392e23210 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1467a91e4faa14fd19e25785db39d4180141b77a","revisions":{"5e4767f913be5af4d205947c100149c4a57e7029":{"kind":"REWORK","_number":1,"created":"2022-06-30 01:30:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/848201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/01/848201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/1"}}},"commit":{"parents":[{"commit":"8672adf57a0fb748e40d0364f734ce57153a8079","subject":"Do not show diffs of image files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/8672adf57a0fb748e40d0364f734ce57153a8079"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-30 01:29:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-30 01:30:14.000000000","tz":540},"subject":"Expose headers option of apache::vhost","message":"Expose headers option of apache::vhost\n\nThe headers option in apache::vhost is required in some case, for\nexample when adding the X-XSS-Protection header. This change allows\ncustomizing the option for the api vhost.\n\nThis change also adds support for request_headers so that both request\nheaders and response headers can customized.\n\nChange-Id: I11274e444c230c62210e2ed5c0f4811f019b16ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/5e4767f913be5af4d205947c100149c4a57e7029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/5e4767f913be5af4d205947c100149c4a57e7029"}]},"branch":"refs/heads/master"},"d1cded33e2a63ee70a37a2288e5836522925d501":{"kind":"REWORK","_number":2,"created":"2022-06-30 01:31:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/848201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/01/848201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/2"}}},"commit":{"parents":[{"commit":"8672adf57a0fb748e40d0364f734ce57153a8079","subject":"Do not show diffs of image files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/8672adf57a0fb748e40d0364f734ce57153a8079"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-30 01:29:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-30 01:30:53.000000000","tz":540},"subject":"Expose headers option of apache::vhost","message":"Expose headers option of apache::vhost\n\nThe headers option in apache::vhost is required in some case, for\nexample when adding the X-XSS-Protection header. This change allows\ncustomizing the option for the api vhost.\n\nThis change also adds support for request_headers so that both request\nheaders and response headers can customized.\n\nChange-Id: I11274e444c230c62210e2ed5c0f4811f019b16ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/d1cded33e2a63ee70a37a2288e5836522925d501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/d1cded33e2a63ee70a37a2288e5836522925d501"}]},"branch":"refs/heads/master"},"1467a91e4faa14fd19e25785db39d4180141b77a":{"kind":"REWORK","_number":3,"created":"2022-06-30 02:03:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/848201/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/01/848201/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/01/848201/3"}}},"commit":{"parents":[{"commit":"8672adf57a0fb748e40d0364f734ce57153a8079","subject":"Do not show diffs of image files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/8672adf57a0fb748e40d0364f734ce57153a8079"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-30 01:29:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-30 02:03:34.000000000","tz":540},"subject":"Expose headers option of apache::vhost","message":"Expose headers option of apache::vhost\n\nThe headers option in apache::vhost is required in some case, for\nexample when adding the X-XSS-Protection header. This change allows\ncustomizing the option for the api vhost.\n\nThis change also adds support for request_headers so that both request\nheaders and response headers can customized.\n\nChange-Id: I11274e444c230c62210e2ed5c0f4811f019b16ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/1467a91e4faa14fd19e25785db39d4180141b77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/1467a91e4faa14fd19e25785db39d4180141b77a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
