)]}'
{"id":"openstack%2Fpuppet-ironic~853089","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fvictoria~I34cff90cbeba1b9b3d043386ee1b47e7039588c2","project":"openstack/puppet-ironic","branch":"stable/victoria","topic":"disk_utils-stable/victoria","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-15 16:31:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I34cff90cbeba1b9b3d043386ee1b47e7039588c2","subject":"Add support for [disk_utils] parameters","status":"MERGED","created":"2022-08-15 03:23:15.000000000","updated":"2022-08-15 19:23:50.000000000","submitted":"2022-08-15 19:22:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853089-disk_utils-stable/victoria","meta_rev_id":"1497977178f2c9cc0400210e9166daa76e3f5832","_number":853089,"virtual_id_number":853089,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-15 19:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-15 08:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-15 14:42:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-15 14:42:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-15 04:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-15 08:08:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-15 14:42:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"67a13b39bf42e741e7509f27c87ba2d647edbe6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 03:23:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8445f7d94e1d1a4a6f5a2bf9a0ba94612f59de29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 04:48:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc5756c82e73447e88e0230ec45b28aa\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/070a36075bf9437e9ccc9c1487ebaa59 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18e26b4416e7493a815e3f4b6b62031b : SUCCESS in 2m 52s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f5ae5369a3542948c82b1a89e57f40f : SUCCESS in 11m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80019e1c42464a1f969e643ac20cd2e2 : SUCCESS in 11m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/368aed840a634d688e4ebd9b862ccd16 : SUCCESS in 11m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03d2c4cf290e43d6ba025e1f4bd1d476 : SUCCESS in 48m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/34b89dfff7fe410982482c66e41a59ff : SUCCESS in 1h 22m 49s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a6e3abc0d954d36a23514106ee841e6 : SUCCESS in 26m 45s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff15bfbb060b4b71bc3d85f3180f9fa5 : SUCCESS in 24m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8e5cab509c54e80ad55a3eec91227fc : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2a40976181daf1a5dc94571c51fb8f9b9d5e94b2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-15 08:08:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"944e1a62af7b7a059e4ddf29ea05a55a61f2f959","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-15 14:42:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"790a90c917e2a5574441820ad6b39b771fd85cd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 14:43:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5aa05ea99d153d532534d9936890f154d13f512d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 16:13:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35ea90fbac064087a95e93d34b7b6119\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3bd88e6baaf40d58e31d4b5b24640a3 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74be34f7e2b946728e81160fb2e02a52 : SUCCESS in 4m 47s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/b16480f58d0649b0a4eb9b7aa7a5b5b1 : SUCCESS in 18m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dcb457d34ff463e8c77e6505c389feb : SUCCESS in 12m 31s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1dbbba3cbd8443788b70c13964072fc1 : FAILURE in 1h 28m 56s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/711bea1d77364a6aaca9efcb32370f1b : SUCCESS in 25m 39s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/011dfb9f28ad4f9eb18d72e0ac1b5a4a : SUCCESS in 18m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d8b60ba12eb4a44982c87d309dd4792 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5468d2aef6ad83d862bc0697b31c0078cd7f51fa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 16:31:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c62be6aa480f23355a8b3f3f2b7c8287d27c9e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 18:10:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a06f49f58335499e89f935b8cc7640a6\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d42e4069c9246d5bb9b74d450f6acbb : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31842fc292d741bcacedc2d3b6089d39 : SUCCESS in 3m 49s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/99ac716368144defb20efcd0cfeebe48 : SUCCESS in 15m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a94fa345b0474930bc31550841b942fd : SUCCESS in 11m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae62bdd220dd40a2b6d18b1f7360cfda : SUCCESS in 11m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd44a44027fe41f5b2d85663963cbb19 : SUCCESS in 48m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a23b185de37441caa517226e4b570561 : SUCCESS in 1h 19m 04s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/9df0fc76bd46425f9009aafa61676dd0 : SUCCESS in 24m 52s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11a4d7d3f5044e56a5148c3a9ef2f9e0 : SUCCESS in 15m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0045838e16664191a53437831051cfc1 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3d5e0b6b4dc6de9ab54020d35a634988461884b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 18:11:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29c8bda5c2accdcec6485daad34fd971eabc1fa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:22:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c2a5934e04a451684e8d4ce34d8a588\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fdce84648f14cbdb877bf6524a688ba : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8746bc14c41472e9f724921dc0b7870 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed5c1b4210314dedb83ee8c34ccc57dd : SUCCESS in 16m 59s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3778dc8343ce46cd9715b89a866a03b3 : SUCCESS in 10m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6268b81d0444e81bfca273b6dc0b8ac : SUCCESS in 1h 11m 32s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ae9629d544e49038f052fdc1ca6d61d : SUCCESS in 22m 33s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efa05d03f12d4dd1bd564f000130f0c5 : SUCCESS in 24m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e1d45868d3d452e915384f624b742a3 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b43ab431c3f71fd2e279ae6d29c3301ac11fefd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:22:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1497977178f2c9cc0400210e9166daa76e3f5832","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:23:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/689b8b54c03b4dfeafa73d8655bf7fc2\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8daf6e9dbeb245839b3503e2ac90b8f8 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f61c5058f0c886c7bbd877568b4f90ebe824329e","revisions":{"f61c5058f0c886c7bbd877568b4f90ebe824329e":{"kind":"REWORK","_number":1,"created":"2022-08-15 03:23:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/853089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/89/853089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/89/853089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/89/853089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/89/853089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/89/853089/1"}}},"commit":{"parents":[{"commit":"16ff3993d3b4f9a98181c670b505bf860e1c0309","subject":"Victoria-only: Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/16ff3993d3b4f9a98181c670b505bf860e1c0309"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-09 05:57:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 03:22:55.000000000","tz":540},"subject":"Add support for [disk_utils] parameters","message":"Add support for [disk_utils] parameters\n\nChange-Id: I34cff90cbeba1b9b3d043386ee1b47e7039588c2\n(cherry picked from commit ba8ff2980f4d8858f35fc2560be8592db17fa45d)\n(cherry picked from commit 6aa42e8486bb8e944d1cbdd1f025950605673bef)\n(cherry picked from commit 21dd14b0543d133025ce5a6b04b849564465071d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f61c5058f0c886c7bbd877568b4f90ebe824329e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f61c5058f0c886c7bbd877568b4f90ebe824329e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
