)]}'
{"id":"openstack%2Fpuppet-ironic~862239","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fwallaby~Ibd0fe69f2a2d0defb648d06e8098712f84de45e6","project":"openstack/puppet-ironic","branch":"stable/wallaby","topic":"debian-ironic-inspector-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-08 08:27:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ibd0fe69f2a2d0defb648d06e8098712f84de45e6","subject":"Debian/Ubuntu: The staging driver package is not available","status":"MERGED","created":"2022-10-24 01:28:33.000000000","updated":"2022-11-08 11:49:00.000000000","submitted":"2022-11-08 11:49:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862239-debian-ironic-inspector-stable/wallaby","cherry_pick_of_change":856208,"cherry_pick_of_patch_set":1,"meta_rev_id":"878b42495890158208d9c1beefafa6e6de8f8d31","_number":862239,"virtual_id_number":862239,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-08 11:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-10-24 15:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-11-04 09:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-11-04 09:09:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-24 04:20:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 15:46:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-11-04 09:09:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f55a3b872f5d0b3dbc60b95812aad44183b4f03a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-24 01:28:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"a93b4182bac80271c87bbe28b90110e700576cbc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-24 01:28:38.000000000","message":"Topic debian-ironic-inspector-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"1736efce3cbf6991bea0c20206440032d19cc378","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-24 01:28:42.000000000","message":"Topic set to debian-ironic-inspector-stable/wallaby","accounts_in_message":[],"_revision_number":1},{"id":"c173673de9cc2ba66ecc60975d82051f4121ec62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 04:20:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d819e7a15744bf881f28425dbc9c247\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bad6eaee79ae46ab9c98a3606da2c3af : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15f2d788063646968a800d572ae5eb21 : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d991de41d934bc98ae975adab297bdc : SUCCESS in 11m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47022765cb33474b88f1870bc274bce8 : SUCCESS in 8m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1746ea2a2ac4dccbeee4e224f91a34d : SUCCESS in 9m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1df6b69e51f4e88865917869fdbdb67 : SUCCESS in 47m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ba767c23db74d16b5acdbd9a11f8ef7 : SUCCESS in 1h 03m 59s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/48b9518b1a744ac38f70bd98b3b416f9 : SUCCESS in 22m 07s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87ed8ea1929046b4b4af418cd650e060 : SUCCESS in 14m 14s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/503c235e4c9348e3897b791e698c93d0 : SUCCESS in 1h 23m 18s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0f89165523e742a98e16b1e587ad348f : SUCCESS in 2h 45m 00s (non-voting)\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bb87f5bf261848599254c357dbb02300 : SUCCESS in 1h 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f0d2535a50c919e6cff0e40d9b00d189707ffc7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-10-24 15:46:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c390004ed1febe26062e86bb8b75d81b0e65d870","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-04 09:09:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e15907855800107e126661f98754d33ee9477831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 09:10:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ac0a0899199669abf52dd5d5905d7dc566b9b12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 10:12:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/56ff06c600ce48e8b8519b0f0b23515e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0959e463de44b039d6e75a7543c5294 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d25b1cb7d50446789ea56bbe4d3d375 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c32796421fbf4f0ba92ceeb4fbd85ff1 : SUCCESS in 13m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0e9b280e10e4e3fbe0f5b1086cba01f : SUCCESS in 8m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c36f0b84826548cd995160dacc82cd3d : SUCCESS in 59m 50s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/2efff920ac704c3cbe3cc57a0e45a49a : SUCCESS in 23m 48s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7582e10aed14d96864015902fa9c687 : SUCCESS in 15m 40s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/72b9377fffe74e5e9fde7b31936814f3 : FAILURE in 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"58eb69550ae21b5975d10080dd10a2a64d861491","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-08 08:27:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"512e64209c4b4fcf57d38633dae35c33505228ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 10:29:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/426253b3a9cf4599a999af44d215dbb7\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3529705d4b0c421c953a5cd2e40a8c6b : SUCCESS in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f92adcb9856b497389ffe9f1113d04dd : SUCCESS in 3m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2d4ff0ce72a458f8029ab7acf6ff39e : SUCCESS in 11m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f462d01e6f34faba73c2c061b9e63f1 : SUCCESS in 7m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/446a45bf0b6b47c5a9f574a22ff14f5c : SUCCESS in 7m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0baab8eaff249f781f1e3296112ea02 : SUCCESS in 53m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/876c1c54de8d43cca3888fd3c659c869 : SUCCESS in 1h 12m 44s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d1feb97c5f34648b6c72ae0f87a23d9 : SUCCESS in 18m 02s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fc3e39d078243749b163476804054fd : SUCCESS in 14m 12s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d5ca6b8d325a448ebe48d1a4be5bc353 : SUCCESS in 1h 12m 49s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/94822ae09a7b471dac9b4763d3ff3b5e : SUCCESS in 1h 56m 45s (non-voting)\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/f8c6a96103cc4866b11ad8c94def4b6f : SUCCESS in 58m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"140ad9c54d2ff5abab9c307e575b953a03f4d033","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 10:30:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5762622e5da33631920728b50314ccee90258b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 11:49:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c307395e6444429968dd5627a6a3aad\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9ff549189b9452a86698067983cad34 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3a9e63c71b547c2badd156ddd78dbfa : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/95dbb02f38e844b494243fa0714f5bf5 : SUCCESS in 12m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d8e707ddc5f42788189de20843dfa0b : SUCCESS in 7m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a09da878c9f44e69bfa527c210c22704 : SUCCESS in 1h 05m 06s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/75949b15f8fd430cb38f4c4025556d54 : SUCCESS in 19m 30s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb501bcb526a4976b3f685e381827ae2 : SUCCESS in 14m 26s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ac5c5494484a4f03a568cb68884a21e7 : SUCCESS in 1h 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"878b42495890158208d9c1beefafa6e6de8f8d31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 11:49:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d15f9a3e8c892bbcaf41093711646039beadba6f","revisions":{"d15f9a3e8c892bbcaf41093711646039beadba6f":{"kind":"REWORK","_number":1,"created":"2022-10-24 01:28:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/862239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/39/862239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/862239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/862239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/862239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/39/862239/1"}}},"commit":{"parents":[{"commit":"079dff9a833da471263f1e9b9438611f040ee9db","subject":"Debian/Ubuntu: Use tftpd-hpa instead of tftpd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/079dff9a833da471263f1e9b9438611f040ee9db"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-03 13:42:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-24 01:28:33.000000000","tz":0},"subject":"Debian/Ubuntu: The staging driver package is not available","message":"Debian/Ubuntu: The staging driver package is not available\n\n... so the deployment should fail the user explicitly requires it.\n\nChange-Id: Ibd0fe69f2a2d0defb648d06e8098712f84de45e6\n(cherry picked from commit a92e08ecc243fab588b813eda06a218d7495b172)\n(cherry picked from commit 1e2207a73339041e4fa412515edb9286ea681501)\n(cherry picked from commit b12fcfe24191b05153336556646ecda1cedc43d3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/d15f9a3e8c892bbcaf41093711646039beadba6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/d15f9a3e8c892bbcaf41093711646039beadba6f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
