)]}'
{"id":"openstack%2Fpuppet-ironic~863363","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fwallaby~I9767330a94f2406a6e76be0f7ce3886478bce3e5","project":"openstack/puppet-ironic","branch":"stable/wallaby","topic":"lp/1966382-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-04 01:22:35.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-11-02 15:58:34.000000000","reason":"Reviewer/Cc was removed"}},"hashtags":[],"change_id":"I9767330a94f2406a6e76be0f7ce3886478bce3e5","subject":"setuptools: Disable auto discovery","status":"MERGED","created":"2022-11-02 15:14:54.000000000","updated":"2022-11-04 01:26:34.000000000","submitted":"2022-11-04 01:26:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863363-lp/1966382-stable/wallaby","cherry_pick_of_change":863213,"cherry_pick_of_patch_set":2,"meta_rev_id":"6404781447dffd5d66421befec0f76df1bd08255","_number":863363,"virtual_id_number":863363,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-04 01:26:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-11-04 01:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-11-02 16:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-11-04 01:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 15:14:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"CC"},{"updated":"2022-11-02 15:24:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 15:56:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2022-11-02 15:58:34.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REMOVED"},{"updated":"2022-11-02 16:21:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"7cb469cd52ad194f547ca1d1004d82d14f48cdbc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 15:14:54.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.\n\nThe following files contain Git conflicts:\n* setup.py","accounts_in_message":[],"_revision_number":1},{"id":"379b675299ee4770816f9b7179947ac8de64d983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 15:24:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60de405a8dda4a91bdd95e36ed867480\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c95faef21c704f3fbaf07505c2c0cd46 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0bab3608f07e6f3d41fb5b5fb019a9ed3c23da0a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 15:56:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cf112761c40fe8dd325f672f084b522f2e312c24","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 15:58:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"30c11158ce7744851ec38fc863e1faee4d4aff41","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 15:58:34.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_32926\u003e.","accounts_in_message":[{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"_revision_number":2},{"id":"15cedb0b4829aa5129705e410fd3b926efd6d5ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 16:04:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b95911ba856e4110a8caa88cb36b986b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b46e5022a6554b1c9066cf52ffceb6c7 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":2},{"id":"e97279824958b805babea99c9673ceca5b531f81","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-02 16:21:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"37571a55c149d6788692f64e708a9caae354fe70","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-04 01:22:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d794114fc1fa7399d6121cfca94714a5e6e86671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:23:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2266c00203e9ff330bb61a1ef0f1d17221dfd80b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:26:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847e44d1bf51469884ca4326a1684f6b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8f2c653448c4066ba3780bd2a28e078 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6404781447dffd5d66421befec0f76df1bd08255","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:26:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"590f7dd626cf7f439631337b982706bf15e66bd2","revisions":{"f0653cfefd5875613940623472b3d91e2beb5176":{"kind":"REWORK","_number":1,"created":"2022-11-02 15:14:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/863363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/63/863363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/63/863363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/63/863363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/63/863363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/63/863363/1"}}},"commit":{"parents":[{"commit":"079dff9a833da471263f1e9b9438611f040ee9db","subject":"Debian/Ubuntu: Use tftpd-hpa instead of tftpd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/079dff9a833da471263f1e9b9438611f040ee9db"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-29 08:39:01.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-02 15:14:54.000000000","tz":0},"subject":"setuptools: Disable auto discovery","message":"setuptools: Disable auto discovery\n\nThe latest release of setuptools 61.0 made a breaking change[1] and\nbecause of this change \u0027pip install\u0027 fails with the following error.\n\n~~~\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027lib\u0027, \u0027spec\u0027, \u0027manifests\u0027, \u0027releasenotes\u0027].\n~~~\n\nUsers that don\u0027t set \u0027packages\u0027, \u0027py_modules\u0027, or configuration\u0027\nare still likely to observe the auto-discovery behavior, which may\nhalt the build if the project contains multiple directories and/or\nmultiple Python files directly under the project root.\n\nTo disable auto discovery, one can do below in setup.py\n\n~~~\nsetuptools.setup(..,packages\u003d[],..)\n~~~\n\nor\n\n~~~\nsetuptools.setup(..,py_modules\u003d[],..)\n~~~\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nNote setup.py is not used to install puppet modules. However it is used\nto generate a release note, thus should be fixed.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I9767330a94f2406a6e76be0f7ce3886478bce3e5\n(cherry picked from commit 58f113cba14ffb94094521a7ef6d5fde6fceb055)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f0653cfefd5875613940623472b3d91e2beb5176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f0653cfefd5875613940623472b3d91e2beb5176"}]},"branch":"refs/heads/stable/wallaby"},"590f7dd626cf7f439631337b982706bf15e66bd2":{"kind":"REWORK","_number":2,"created":"2022-11-02 15:56:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/863363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/63/863363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/63/863363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/63/863363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/63/863363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/63/863363/2"}}},"commit":{"parents":[{"commit":"079dff9a833da471263f1e9b9438611f040ee9db","subject":"Debian/Ubuntu: Use tftpd-hpa instead of tftpd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/079dff9a833da471263f1e9b9438611f040ee9db"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-29 08:39:01.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-02 15:56:27.000000000","tz":540},"subject":"setuptools: Disable auto discovery","message":"setuptools: Disable auto discovery\n\nThe latest release of setuptools 61.0 made a breaking change[1] and\nbecause of this change \u0027pip install\u0027 fails with the following error.\n\n~~~\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027lib\u0027, \u0027spec\u0027, \u0027manifests\u0027, \u0027releasenotes\u0027].\n~~~\n\nUsers that don\u0027t set \u0027packages\u0027, \u0027py_modules\u0027, or configuration\u0027\nare still likely to observe the auto-discovery behavior, which may\nhalt the build if the project contains multiple directories and/or\nmultiple Python files directly under the project root.\n\nTo disable auto discovery, one can do below in setup.py\n\n~~~\nsetuptools.setup(..,packages\u003d[],..)\n~~~\n\nor\n\n~~~\nsetuptools.setup(..,py_modules\u003d[],..)\n~~~\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nNote setup.py is not used to install puppet modules. However it is used\nto generate a release note, thus should be fixed.\n\nConflicts:\n\tsetup.py\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I9767330a94f2406a6e76be0f7ce3886478bce3e5\n(cherry picked from commit 58f113cba14ffb94094521a7ef6d5fde6fceb055)\n(cherry picked from commit 0d8437c73fb8afef6da829f7728872dd21ff37a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/590f7dd626cf7f439631337b982706bf15e66bd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/590f7dd626cf7f439631337b982706bf15e66bd2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
