)]}'
{"id":"openstack%2Fpuppet-ironic~871900","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fyoga~I49a295dd23eedd758bc9bb2e8a7ff9e740cb2c42","project":"openstack/puppet-ironic","branch":"stable/yoga","topic":"bug/2003681","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-06 17:43:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I49a295dd23eedd758bc9bb2e8a7ff9e740cb2c42","subject":"inspector: Fix wrong group of policy file","status":"MERGED","created":"2023-01-27 01:54:03.000000000","updated":"2023-02-06 17:43:27.000000000","submitted":"2023-02-06 17:43:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871900-bug/2003681","cherry_pick_of_change":871553,"cherry_pick_of_patch_set":1,"meta_rev_id":"c8b1caf9f78a3f3105b2baec17f289ec2f27314d","_number":871900,"virtual_id_number":871900,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-06 17:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-02-02 15:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-02-06 16:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-02-06 16:24:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 04:03:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-02 15:43:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-02-06 16:24:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9e544887c581feb6f740d5fba4aed95af843c771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-27 01:54:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"16b9c4c7f320c8a754382611cf32341a929d6018","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-27 01:54:11.000000000","message":"Topic bug/2003681-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"974e616714b7ac0af21f7f3919ec6535a6e0dc69","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-27 01:54:15.000000000","message":"Topic set to bug/2003681","accounts_in_message":[],"_revision_number":1},{"id":"71ed17ca92b9c7e247622d39753f619214308b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 04:03:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ec17e8a47ce4f95b3a45c1e27b01053\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec632be46f3341cea5d3208687d6d36e : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acc45ae4f0de4b8d8d0287c29ee1740f : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9863e3c700a4a60be30c0c23b41eb45 : SUCCESS in 2m 54s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/179f2001e4ef48b49ec08e1742e5b272 : SUCCESS in 13m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee94df8c32b94cfe85429d941a097085 : SUCCESS in 11m 01s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f42898ccabf84cffb7368b6fc961f329 : SUCCESS in 9m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e30f456eac7843968fe7e189ff55ef0b : SUCCESS in 9m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/704f1e66cbf04f12a99ad0299e9cf318 : FAILURE in 1h 15m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15f82efe552f4686a2ff55f1d2994a31 : TIMED_OUT in 2h 07m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b947cae9e25b4132b33bebfd49b2035e : SUCCESS in 1h 16m 43s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b52b6dcaa9e54b288cd7383886a6c56d : SUCCESS in 14m 28s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c18b82caec1c4741adcf775953cba07c : SUCCESS in 27m 25s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/7959b4db8a0946e891b63c50fcfa1989 : SUCCESS in 20m 46s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/959b19c5aa4e41d1b13e571f2800fb62 : SUCCESS in 15m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f55bdda5641f4341a81e9319d44a8969 : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0fff9b9e90dfde92f8b2d8683693e1124f547fac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-01 01:52:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b2cf3e1f8b084753af2eb48f416515d343a11c93","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-01 01:52:47.000000000","message":"Topic bug/2003681-stable/yoga removed","accounts_in_message":[],"_revision_number":2},{"id":"abea93fb7bbb274a03eb14cb7af4a7c3870567ce","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-01 01:52:52.000000000","message":"Topic set to bug/2003681","accounts_in_message":[],"_revision_number":2},{"id":"a5eb6b140d5e3129a8c34970dd4a17299754faeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 03:55:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94998f7871a74df0b05119b9917caa26\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e14e5251432f49049ebd6701040267c3 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b49a20b3c09413a875a6a7676182efb : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce5e8b01e97948a18915a2022b3d350c : SUCCESS in 2m 50s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f13690738a0547bcb38de0b9fd513a5e : SUCCESS in 14m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc42a2b3e10f42c2bbe724c5a6392270 : SUCCESS in 13m 24s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8123680465947a19f1e81e5f4d076c5 : SUCCESS in 10m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ec462a797af46f38e813de2919bd296 : SUCCESS in 11m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d43cde8b2a174b39874faa0aca34a161 : FAILURE in 53m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/52288df741bf486e8ca35e92c2295bef : SUCCESS in 1h 54m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b53fa8184eeb47d4a3a9aaed7f12db6d : SUCCESS in 1h 11m 30s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd73fbbb46374761a6101443f5299394 : SUCCESS in 14m 55s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/81b83d11e10742c3827aa01e0d1bd300 : SUCCESS in 24m 53s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0c71124c9294dadbb07b519b04c7ba9 : SUCCESS in 20m 50s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d661441ed144f94bbc5794445947506 : SUCCESS in 16m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08d9456637104acea001d9273451363f : SUCCESS in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d801c6a8f4fc886260297f6b520d01074453b8c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-02-02 15:43:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"073ec6d33b72318976d4133130ab1c38db39c2ae","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-06 16:24:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"88f154a3f4d6daa68678b9e8fee467755b77f84e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 16:25:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f12d76ab7e6eb481aa1236f078a088243b2af3ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 17:43:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cbe5fd837104bceac68758c412b46a2\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f816b87bed54ec4be648c61403c987e : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67cea208cb2c45eea06764f979be8700 : SUCCESS in 3m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9d39754351c48e3bfaa35f6de7a5ed2 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcb3c7d6475e46d79e6217018264f122 : SUCCESS in 12m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c93af31f5cd849599ee9f03c7c8adcc8 : SUCCESS in 13m 23s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25670c4dc7bf47f8bbd6efb1d7c0428f : SUCCESS in 9m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4cfadf813b4b4c0f8520b6a5c4cb080a : SUCCESS in 1h 10m 27s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/24fd9e53536b43b2a9eecc6fcb21f2d3 : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c8b1caf9f78a3f3105b2baec17f289ec2f27314d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 17:43:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2745385b26c921ca5469d43544d129c57b35e31e","revisions":{"bf624662f85e3ffe11618dc6dc04706a28e91aa0":{"kind":"REWORK","_number":1,"created":"2023-01-27 01:54:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/871900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/00/871900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/00/871900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/00/871900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/00/871900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/00/871900/1"}}},"commit":{"parents":[{"commit":"f63b1bd2d36d058c40720d6ca45ba017ec25c812","subject":"Ubuntu: Use ipxe.efi instead of snponly.efi in Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f63b1bd2d36d058c40720d6ca45ba017ec25c812"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 06:04:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-27 01:54:03.000000000","tz":0},"subject":"inspector: Fix wrong group of policy file","message":"inspector: Fix wrong group of policy file\n\nThe ironic-inspector service is launched using the separate user/group\n(ironic-inspector) so the file should be owned by that group instead of\nthe ironic group.\n\nCloses-Bug: #2003681\nChange-Id: I49a295dd23eedd758bc9bb2e8a7ff9e740cb2c42\n(cherry picked from commit 439927462fe651a8d4ee7788e5f0b00ed0f64d09)\n(cherry picked from commit 3802e37495513b9c306d57a90159eac028611d59)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/bf624662f85e3ffe11618dc6dc04706a28e91aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/bf624662f85e3ffe11618dc6dc04706a28e91aa0"}]},"branch":"refs/heads/stable/yoga"},"2745385b26c921ca5469d43544d129c57b35e31e":{"kind":"NO_CHANGE","_number":2,"created":"2023-02-01 01:52:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/871900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/00/871900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/00/871900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/00/871900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/00/871900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/00/871900/2"}}},"commit":{"parents":[{"commit":"f63b1bd2d36d058c40720d6ca45ba017ec25c812","subject":"Ubuntu: Use ipxe.efi instead of snponly.efi in Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f63b1bd2d36d058c40720d6ca45ba017ec25c812"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 06:04:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-01 01:52:41.000000000","tz":0},"subject":"inspector: Fix wrong group of policy file","message":"inspector: Fix wrong group of policy file\n\nThe ironic-inspector service is launched using the separate user/group\n(ironic-inspector) so the file should be owned by that group instead of\nthe ironic group.\n\nCloses-Bug: #2003681\nChange-Id: I49a295dd23eedd758bc9bb2e8a7ff9e740cb2c42\n(cherry picked from commit 439927462fe651a8d4ee7788e5f0b00ed0f64d09)\n(cherry picked from commit 3802e37495513b9c306d57a90159eac028611d59)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/2745385b26c921ca5469d43544d129c57b35e31e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/2745385b26c921ca5469d43544d129c57b35e31e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
