)]}'
{"id":"openstack%2Fpuppet-ironic~882902","triplet_id":"openstack%2Fpuppet-ironic~master~Ibe3e54ece03f112d5fa93c22b402116050e65402","project":"openstack/puppet-ironic","branch":"master","topic":"python-systemd-package-name","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2023-05-11 07:51:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe3e54ece03f112d5fa93c22b402116050e65402","subject":"params.pp: Fix deb systemd_python_package","status":"MERGED","created":"2023-05-10 22:23:06.000000000","updated":"2023-05-11 07:51:24.000000000","submitted":"2023-05-11 07:51:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882902-python-systemd-package-name","meta_rev_id":"e916f3e0de9aa332826c57814b382292d9af36a2","_number":882902,"virtual_id_number":882902,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-11 07:51:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-05-11 01:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-05-11 06:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-05-11 06:36:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-10 23:28:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-11 01:44:34.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-11 06:36:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"90f5c30f889c3049cd84a0ffa3700a9e5a71539d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-05-10 22:23:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e861298465eea80308a4ee15eaf3a2fcebb12b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 23:28:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d95e74c154124055af1bf2bed1d77a82\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca05d62adef341c793fe4473e489aefc : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ccaee7673db40b096c80975e4372c55 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a895e4125614f62947b54ce2080f631 : SUCCESS in 12m 52s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d06f152257540b49e2874a5f3b8af6f : SUCCESS in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1856749f1b3e4ffdbd9530c2316572ac : SUCCESS in 9m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b68e82f94754f1797a7d0b10abd831c : SUCCESS in 52m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5264f460ad7844d3a96ff7620318d9fa : FAILURE in 1h 04m 14s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8a92c773f26454c93000d17bada554d : SUCCESS in 16m 37s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03d1e642f16b43019f3bc788f31c020b : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d325b300609f4041a0f512288322438f : SUCCESS in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71450cf7f6d6d3a35c5687a660b2d254ab56aebf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-11 01:44:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d1420008610efe0605048f77f929b23afea37be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 03:08:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ac55005538b48af82402f61dae7b9fb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff478f2d927e444ab57ef045c3f04ee7 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/227a8db393524d849bd790755bf00c47 : SUCCESS in 2m 19s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a398ac47873243b383a626e4a8021600 : SUCCESS in 12m 16s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d591edccd207468c89a6e444ff814f6b : SUCCESS in 7m 11s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91b0e097b8a04a4b93b933dada10adcf : SUCCESS in 7m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55761c6f45b84a4fb6a8685b4f2dd4a8 : SUCCESS in 1h 00m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b399945dfe0496bad6b63009dbb41dd : SUCCESS in 1h 22m 32s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbfc35f268e84757a5d3fb8ff57559f9 : SUCCESS in 18m 38s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8db45c46ebfe4c6e910b24407985a5c8 : SUCCESS in 13m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1264fe5e546f4128a2a631382c71084f : SUCCESS in 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05163edc773b31b2a9ae816e0be011b5e50c71f7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-05-11 06:36:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e7e1c13e3c44c46f87abba232b338caa5093be0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 06:37:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43b165c306dceea74e73e71b77ccf397a762d97f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 07:51:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d26519989db24ffb902f65a5fd90f423\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cbf460f62cd46d4a842023744778503 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aed2dc2c9e4344afbcebd2446e27b2cc : SUCCESS in 3m 08s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/05f42132aad74462920d58ae61179ec9 : SUCCESS in 13m 30s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fd15ac3b5534fabacfe5bf79d529b1b : SUCCESS in 8m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3f124a0fb584f5c8b795bb7a68abf63 : SUCCESS in 1h 08m 54s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/be1dcb484d6746b9aa2bcd84b2a56663 : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e916f3e0de9aa332826c57814b382292d9af36a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 07:51:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"643b3d1473c29c4eda0c85899384c697902c4c56","revisions":{"643b3d1473c29c4eda0c85899384c697902c4c56":{"kind":"REWORK","_number":1,"created":"2023-05-10 22:23:06.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/02/882902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/02/882902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/02/882902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/02/882902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/02/882902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/02/882902/1"}}},"commit":{"parents":[{"commit":"5219b24dc9b845b9c4ccedf7ac4394b7ea381a9d","subject":"Bump upper version of puppetlabs-apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/5219b24dc9b845b9c4ccedf7ac4394b7ea381a9d"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-05-10 22:20:50.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-05-10 22:22:55.000000000","tz":120},"subject":"params.pp: Fix deb systemd_python_package","message":"params.pp: Fix deb systemd_python_package\n\nThis fixes a wrong package name in Debian and Ubuntu\nin params.pp:\npython-systemd -\u003e python3-systemd\n\nChange-Id: Ibe3e54ece03f112d5fa93c22b402116050e65402\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/643b3d1473c29c4eda0c85899384c697902c4c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/643b3d1473c29c4eda0c85899384c697902c4c56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
