)]}'
{"id":"openstack%2Fpuppet-keystone~1004746","triplet_id":"openstack%2Fpuppet-keystone~master~I3b66b1ea5c159ced8eefb43fcd2c118cc8c920df","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 16:22:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3b66b1ea5c159ced8eefb43fcd2c118cc8c920df","subject":"Deprecate support for [database] mysql_enable_nbd","status":"MERGED","created":"2026-09-09 02:46:58.000000000","updated":"2026-09-10 16:23:51.000000000","submitted":"2026-09-10 16:22:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004746","meta_rev_id":"bb01bdbbc1c7f098977e54d06935b1b6200ab2d8","_number":1004746,"virtual_id_number":1004746,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 16:22:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2026-09-10 10:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2026-09-10 14:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2026-09-10 14:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 04:49:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 10:54:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-09-10 14:49:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"eab0ba56808df14cf0f2ad21ed5a2e718d370495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 02:46:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"502ef2db495e97ff224c26833869f13b9b6dbb23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 04:49:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b0d782ea82f412c9628aa565bdb25a1\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f0379a4dece47d1a1191b8640bc96f4 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a71729616dc145cebb95d3606d987cef : SUCCESS in 3m 14s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/05601ca03c734046879af51fa6bb378c : FAILURE in 6m 16s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bc3bd8ccd0a46d5aee696569cb01847 : FAILURE in 6m 15s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63e241e50f7a40899ddee4e44d133f35 : FAILURE in 7m 03s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c9e2f70bbd84a3baa5b188c19ac718d : SUCCESS in 31m 47s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf547fc926764eec955be3c9c1007280 : FAILURE in 1h 17m 37s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/515bbb921d72406685877bb50b99de61 : FAILURE in 1h 21m 17s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db3e37fb6bc34d9381345d074109fd40 : SUCCESS in 29m 03s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d333547a800b4f3ca6dcc1ec14afc104 : FAILURE in 36m 17s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a81b4fda06544e3c89cf980e42dfe6f0 : SUCCESS in 1h 04m 19s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/554a36b1fca842359b3caab3efbc8283 : SUCCESS in 1h 15m 22s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/282cace3982e4769b079a54d68009f45 : SUCCESS in 56m 03s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac734118f8c9489c80f4b845bb23602f : SUCCESS in 34m 52s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b742fbd1928d4395adda9b68ab65f64c : SUCCESS in 1h 02m 56s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70e299a563ef4e49991b1b558f5e66a3 : SUCCESS in 1h 08m 47s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f05edc68471d43d2a688cf92b79d72f0 : SUCCESS in 1h 18m 42s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fc952e7514b4b9181bd5ab091886b1a : SUCCESS in 58m 07s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/86a12b9c32f84c23aa265974945009d3 : SUCCESS in 1h 09m 02s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f87f5eb059545f68c3052c141a7a168 : SUCCESS in 58m 03s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b1e410714234b0297ce58f84a8edf54 : FAILURE in 8m 53s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c82d5c9fb82f4259a1c4dcb2e3bba67f : SUCCESS in 18m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fde469fc096249aeafa3078890fbb2b8 : SUCCESS in 25m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ec67f4a515f4c6ca4a35827fca17254 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6cbad8ad335df4c4a1ba8a596cd0a41d92cd0e48","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 16:20:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"345ab41a57302a40c1d668d5c6c05b3d16dd92c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 17:56:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d690076d4784bfe8de1f0506233f222\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebf9c660ad5043649803ef2d8767324e : SUCCESS in 4m 32s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4f7c9a0516a43eda982e3fefe3c57a8 : SUCCESS in 3m 52s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/4a1d49b524324af1b1a3371b00cbfb11 : SUCCESS in 14m 49s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57a7f7ceabe840678f1639d1fd74bb13 : SUCCESS in 13m 55s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e943288138a4d5f8fd8d27830614b4c : SUCCESS in 14m 39s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0c38009678943e08d6300a9e0653c49 : SUCCESS in 43m 37s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1216441e4759472a8b8ae25ce6cabee2 : FAILURE in 50m 01s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5cc579837884b30b5bdb17b4b4ce2f2 : FAILURE in 1h 21m 17s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33ce0c17858e4374a97c962e7305f75e : SUCCESS in 51m 30s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5436d2dd9e634df088c41b8afa49d654 : FAILURE in 55m 48s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/498d06285c7c48f9b9630a9e7909dba0 : SUCCESS in 1h 13m 41s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bc48a8e33ef49bc886e3688f2955001 : SUCCESS in 1h 24m 31s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb87ebd62bc342fbbde4822b958fa908 : SUCCESS in 1h 01m 25s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c50ba4a8bb13468da004d397ed3f9208 : SUCCESS in 1h 09m 37s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fa8ed63aab14891a45a18b3682ab653 : SUCCESS in 1h 05m 21s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a62ff86cb48433fbb8b1d4aebf4cf93 : SUCCESS in 55m 29s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d554280e7d6b40218c8cf848ce679752 : SUCCESS in 1h 25m 08s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c40e8c4d3a7f48a08e532189656a1cc5 : SUCCESS in 1h 05m 51s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/63bbd3a558704849a7ec0d34b305d572 : SUCCESS in 56m 05s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/574c1fa2ccc74b7588d154f371d5f4e8 : SUCCESS in 57m 56s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4849a94a26c548938cee2f179f7d1428 : SUCCESS in 15m 02s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1634ed3dc688462f96df638dc0bf64e1 : SUCCESS in 24m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21d643add00840a7a2a6d344326394f2 : SUCCESS in 24m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f6319ee17e54de2b2e6901effc1ee92 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"99cafa271d8460df3fa5500dc0ccb90b4d6b341f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-10 10:54:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d8d81fcbed3db1de6a1b914f9b37d6731f532c1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-09-10 14:49:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2277f15b997a5a11c3a268665d7b03aa6b768ba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:57:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c357154d4736e01a111b8e534cbeb1e322bf4fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:22:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b428e5c29ed48ad9dafa821897dd98c\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40bc5ee72c9a40b598485355d07d6926 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6869431216d4e738e0046e5295d093b : SUCCESS in 2m 57s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/094a319c4e094f30a4a538e103ea061c : SUCCESS in 11m 17s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/228547fef7eb4a7f8e6b64dbd48793ef : SUCCESS in 19m 29s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce3ef5634d3145b48fd46f3a4fae6ea2 : SUCCESS in 1h 07m 18s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71ec823225e849679edbdcedc9ebffee : SUCCESS in 1h 07m 58s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b148d3a3f30437c948ab3e7187f02c0 : SUCCESS in 51m 39s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc09b0558a654baf883d742e79268485 : SUCCESS in 1h 01m 29s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/277bfd03ad9840d6b03412fc7c1b18d8 : SUCCESS in 1h 10m 34s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bd94e9fa85c4b149d682cadc95bb60a : SUCCESS in 34m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbc9c298ac684ffd828abdfb14e688a3 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8418a862d7cc3a13d5c07d4814ae4bad1346e19f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:22:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bb01bdbbc1c7f098977e54d06935b1b6200ab2d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:23:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b53ff3b6b7534987a1b60109a076bacd\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/437bfc7fa57544719afbabfd0589a2e9 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6430e0c9be155d962934df9662924d646cfcc46","revisions":{"c6430e0c9be155d962934df9662924d646cfcc46":{"kind":"REWORK","_number":1,"created":"2026-09-09 02:46:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/1004746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/46/1004746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/46/1004746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/46/1004746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/46/1004746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/46/1004746/1"}}},"commit":{"parents":[{"commit":"02f911ae4c52262bd8d08faac75b67fba783cea6","subject":"Remove old upgrade warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/02f911ae4c52262bd8d08faac75b67fba783cea6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:46:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:46:48.000000000","tz":540},"subject":"Deprecate support for [database] mysql_enable_nbd","message":"Deprecate support for [database] mysql_enable_nbd\n\nDepends-on: https://review.opendev.org/1004735\nChange-Id: I3b66b1ea5c159ced8eefb43fcd2c118cc8c920df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/c6430e0c9be155d962934df9662924d646cfcc46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/c6430e0c9be155d962934df9662924d646cfcc46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
