)]}'
{"id":"openstack%2Fpuppet-keystone~185441","triplet_id":"openstack%2Fpuppet-keystone~master~I43657e9ba46f07c316a9e67a134e30aa3a44a867","project":"openstack/puppet-keystone","branch":"master","hashtags":[],"change_id":"I43657e9ba46f07c316a9e67a134e30aa3a44a867","subject":"Add support for Fernet Tokens","status":"MERGED","created":"2015-05-25 18:48:32.000000000","updated":"2015-06-11 17:45:49.000000000","submitted":"2015-06-11 17:45:48.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f5a89ca0b80508934e28893e4b08a910f2e0c889","_number":185441,"virtual_id_number":185441,"owner":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2015-06-11 17:45:48.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":0,"_account_id":10540,"name":"Clayton O\u0027Neill","email":"clayton.oneill@charter.com","username":"coneill"},{"value":0,"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2015-06-11 04:32:45.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":1,"date":"2015-06-11 08:51:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":2,"date":"2015-06-11 17:38:38.000000000","_account_id":10540,"name":"Clayton O\u0027Neill","email":"clayton.oneill@charter.com","username":"coneill"},{"value":0,"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10540,"name":"Clayton O\u0027Neill","email":"clayton.oneill@charter.com","username":"coneill"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":1,"date":"2015-06-11 17:38:38.000000000","_account_id":10540,"name":"Clayton O\u0027Neill","email":"clayton.oneill@charter.com","username":"coneill"},{"value":0,"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},{"_account_id":10540,"name":"Clayton O\u0027Neill","email":"clayton.oneill@charter.com","username":"coneill"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-05-26 08:07:54.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"REVIEWER"},{"updated":"2015-05-26 20:39:09.000000000","updated_by":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"reviewer":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"state":"REVIEWER"},{"updated":"2015-05-29 04:06:14.000000000","updated_by":{"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},"reviewer":{"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},"state":"REVIEWER"},{"updated":"2015-05-29 15:15:00.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2015-06-11 04:32:45.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2015-06-11 08:51:03.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2015-06-11 17:38:38.000000000","updated_by":{"_account_id":10540,"name":"Clayton O\u0027Neill","email":"clayton.oneill@charter.com","username":"coneill"},"reviewer":{"_account_id":10540,"name":"Clayton O\u0027Neill","email":"clayton.oneill@charter.com","username":"coneill"},"state":"REVIEWER"},{"updated":"2015-06-11 17:45:48.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"33862497532131791d0c5a9513e529f112537a57","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-05-25 18:48:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af4e750655d7a8de121acc62e957e46387e4d3d5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-25 19:02:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-lint/28823f9/ : FAILURE in 1m 29s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-syntax/e83f083/ : SUCCESS in 3s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-unit-3.3/f2eab7c/ : SUCCESS in 2m 24s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-unit-3.4/33c0427/ : SUCCESS in 2m 21s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-unit-3.6/cb6efc5/ : SUCCESS in 3m 08s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-unit-3.7/dc78f15/ : SUCCESS in 2m 16s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-unit-4.0/e54264f/ : SUCCESS in 2m 09s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/fde6ee3/ : SUCCESS in 12m 57s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/c149f75/ : SUCCESS in 13m 35s (non-voting)\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/1/check/gate-puppet-keystone-puppet-syntax-future/2163755/ : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2fffdbe74203c736a579542404afbf5927cb5758","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-05-25 19:03:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"be3a11c5fe1888566403f4f1daf14a8bdae07795","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-25 19:18:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-lint/8a7830e/ : SUCCESS in 2m 03s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-syntax/8f1a175/ : SUCCESS in 9s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-unit-3.3/c654cb0/ : SUCCESS in 3m 19s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-unit-3.4/0e7a13b/ : SUCCESS in 2m 22s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-unit-3.6/32e2df2/ : SUCCESS in 2m 37s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-unit-3.7/4c95e58/ : SUCCESS in 2m 44s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-unit-4.0/bfbe45b/ : SUCCESS in 3m 07s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/9737096/ : SUCCESS in 12m 36s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/2e1f6d5/ : SUCCESS in 14m 59s (non-voting)\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/2/check/gate-puppet-keystone-puppet-syntax-future/a52c11c/ : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b73a5e036bae041905a8e39bd6cb8b5e95e5e301","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-25 20:17:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (check-tripleo pipeline).\n\n- check-tripleo-ironic-overcloud-f20puppet-nonha http://logs.openstack.org/41/185441/2/check-tripleo/check-tripleo-ironic-overcloud-f20puppet-nonha/2f2e9ea/ : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":2},{"id":"bdf293f7bb4c7d49ee8e979abf7b9ffc2dec180e","author":{"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},"date":"2015-05-26 02:07:14.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nLooking good! A few comments inline.","accounts_in_message":[],"_revision_number":2},{"id":"8d1daa2f0d969b1b0f93921c26f7947fc0f19bbd","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-05-26 03:13:12.000000000","message":"Patch Set 2:\n\n1) fixed\n3) fixed.\n\nAs for 2) I thought about this alot. I had it undef also but in reality that makes no sense. I need to have a value here so I can set the creates section and also test it. If I don\u0027t default it to the keystone default then I need to have some kind of backup  here.","accounts_in_message":[],"_revision_number":2},{"id":"2219c479e4b7986e3697ec68f40bb64eef3769e9","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-05-26 03:13:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"808e69162ebe884b1378247988d31df0150395e3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-26 03:24:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-lint/3efa994/ : SUCCESS in 1m 24s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-syntax/ec58b69/ : FAILURE in 3s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-unit-3.3/92546e7/ : FAILURE in 3m 45s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-unit-3.4/7434c0f/ : FAILURE in 4m 14s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-unit-3.6/63de72c/ : FAILURE in 3m 36s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-unit-3.7/91f10e4/ : FAILURE in 3m 19s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-unit-4.0/cb4b721/ : FAILURE in 3m 54s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/ec4e488/ : FAILURE in 10m 59s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/e386593/ : FAILURE in 11m 08s (non-voting)\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/3/check/gate-puppet-keystone-puppet-syntax-future/4d1cf6e/ : FAILURE in 1m 20s","accounts_in_message":[],"_revision_number":3},{"id":"e287a8ccdb72a65dcf5696af93a2abb5d2c0cb0c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-26 04:19:02.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- check-tripleo-ironic-overcloud-f20puppet-nonha http://logs.openstack.org/41/185441/3/check-tripleo/check-tripleo-ironic-overcloud-f20puppet-nonha/e1a7a1b/ : FAILURE in 1h 05m 25s","accounts_in_message":[],"_revision_number":3},{"id":"77a4ebc750487ba205a3df3b0f9230fbe582965d","author":{"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},"date":"2015-05-26 04:34:20.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nOk, makes sense on the #2 thing.\n\nIn researching this stuff more tonight, just one more comment.","accounts_in_message":[],"_revision_number":3},{"id":"1ba224e69f5c92f5eb096c45d02fb9cbef24b378","author":{"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},"date":"2015-05-26 04:37:31.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"540e8b4fa675653e485a5fe57f59096565a8ffdf","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-05-26 19:37:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"16af6db67828d14248a3276238d9b97a434c7a80","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-05-26 19:40:20.000000000","message":"Patch Set 3:\n\nThis change does not enable fernet tokens, it enables the fernet token setup, hence the name of the flag. You would only run this on one node in a cluster. Therefore I\u0027m not forcing the token provider for you.","accounts_in_message":[],"_revision_number":3},{"id":"234c3e001ce7ba5f6a4d921347a0ef0a089eac7a","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2015-05-26 19:49:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c9656593e2910222a4038e53ff26fa2502330333","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-26 19:54:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-lint/84b7a07/ : SUCCESS in 2m 12s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-syntax/fc71719/ : SUCCESS in 10s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-unit-3.3/adae2a3/ : SUCCESS in 2m 41s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-unit-3.4/7e11eae/ : SUCCESS in 3m 38s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-unit-3.6/c63b32b/ : SUCCESS in 3m 18s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-unit-3.7/115cc69/ : SUCCESS in 3m 46s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-unit-4.0/ef04815/ : SUCCESS in 3m 17s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/b90338d/ : SUCCESS in 14m 24s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/2706815/ : SUCCESS in 12m 19s (non-voting)\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/4/check/gate-puppet-keystone-puppet-syntax-future/b5c5788/ : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":4},{"id":"fa030a0ed4216c6c16aab814764b65d5edd413a5","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-05-26 20:27:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7a0fff1beae1412fe8c80bba86c570fb62096e75","author":{"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},"date":"2015-05-26 20:38:25.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"030127e59c628447688492d165f76c37f238a3fa","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2015-05-26 20:39:14.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f5cc0bf8766d9b4fbab3063afee1a7c015e82409","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-26 20:42:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-lint/6a1bbb3/ : SUCCESS in 2m 13s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-syntax/f4da25b/ : SUCCESS in 11s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-unit-3.3/049325d/ : SUCCESS in 2m 48s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-unit-3.4/9797949/ : SUCCESS in 2m 51s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-unit-3.6/ca59693/ : SUCCESS in 2m 51s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-unit-3.7/78f417a/ : SUCCESS in 2m 50s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-unit-4.0/fa061e2/ : SUCCESS in 3m 06s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/eb47f04/ : SUCCESS in 13m 40s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/82a7f70/ : SUCCESS in 13m 04s (non-voting)\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/5/check/gate-puppet-keystone-puppet-syntax-future/39832f8/ : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":5},{"id":"710c98ae87a6144767801a4341ee3380d1a24954","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-26 21:42:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (check-tripleo pipeline).\n\n- check-tripleo-ironic-overcloud-f20puppet-nonha http://logs.openstack.org/41/185441/5/check-tripleo/check-tripleo-ironic-overcloud-f20puppet-nonha/4e9441e/ : SUCCESS in 1h 15m 22s","accounts_in_message":[],"_revision_number":5},{"id":"1d5832e8205dfba1e49f44dedca7c44bcb1c381d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2015-05-28 18:00:30.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e9db8c10bbed4bd02450a3749ac96de8abac9c01","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-05-29 03:49:32.000000000","message":"Patch Set 5:\n\nColleen, should we bother with the check at all? We don\u0027t typically do that anywhere else, but its simple when its 1 line rather than several.","accounts_in_message":[],"_revision_number":5},{"id":"3d33c73f32a601a77dce28740688adf7d9d6076f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2015-05-29 03:55:08.000000000","message":"Patch Set 5:\n\nI\u0027m fine with omitting it.","accounts_in_message":[],"_revision_number":5},{"id":"d0fd4a9e308e65fde6834d1bcbeef174ca1c11ef","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-05-29 03:59:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4cd78474ef69ee4835847887000efb7a2448c229","author":{"_account_id":9060,"name":"Mike Dorman","email":"dorm@dorm.org","username":"gd-mdorman"},"date":"2015-05-29 04:06:14.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c221b1ce0872fe7e2c1f5b20061b73c9f094dfed","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-29 05:14:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (check-tripleo pipeline).\n\n- check-tripleo-ironic-overcloud-f20puppet-nonha http://logs.openstack.org/41/185441/6/check-tripleo/check-tripleo-ironic-overcloud-f20puppet-nonha/8d24993/ : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":6},{"id":"6c1252d82ca3e44fbcbedba0847d09d2ccb6a628","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-29 06:22:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-lint/a95a6f9/ : SUCCESS in 1m 20s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-syntax/697c5d0/ : SUCCESS in 3s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-3.3/00ba863/ : SUCCESS in 2m 33s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-3.4/9f7542c/ : SUCCESS in 2m 26s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-3.6/003fcb0/ : SUCCESS in 2m 30s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-3.7/f5dd188/ : SUCCESS in 2m 30s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-4.0/3995880/ : SUCCESS in 2m 23s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/4e91a29/ : SUCCESS in 13m 16s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/bf404e5/ : FAILURE in 13m 00s (non-voting)\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-syntax-future/98c9e2f/ : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":6},{"id":"c09d233d1c00372e7a9336efcebcf433719d9cce","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2015-05-29 15:15:00.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5a3aaf1becfc4ba3fd5288c30aeabce339e24a0f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-29 16:31:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (check-tripleo pipeline).\n\n- check-tripleo-ironic-overcloud-f20puppet-nonha http://logs.openstack.org/41/185441/6/check-tripleo/check-tripleo-ironic-overcloud-f20puppet-nonha/8de6a77/ : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":6},{"id":"e035f74c9f0cf2a631b1f877c53caf5349b8f562","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-29 17:53:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-lint/9fbed7c/ : SUCCESS in 2m 04s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-syntax/7372fa0/ : SUCCESS in 2s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-3.3/6df5c20/ : SUCCESS in 2m 59s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-3.4/a8feb47/ : SUCCESS in 3m 21s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-3.6/0be4ed6/ : SUCCESS in 3m 17s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-3.7/42751ed/ : SUCCESS in 3m 20s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-unit-4.0/e580363/ : SUCCESS in 2m 53s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/2fbbe33/ : SUCCESS in 12m 26s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/d9f784f/ : SUCCESS in 12m 57s (non-voting)\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/6/check/gate-puppet-keystone-puppet-syntax-future/8e73979/ : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":6},{"id":"d42839f26ddc4c584221753d8eb0f5030dcb80ee","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2015-05-29 18:07:36.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"d059df335361833f7c90c3980113a9807b26a41d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2015-06-01 12:03:50.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"0c8d0104dc76c50699d8a458ce20b64348e9c0a1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2015-06-07 14:19:36.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nThanks for this work and all your testing you did around!","accounts_in_message":[],"_revision_number":6},{"id":"617083035af8cbb7470174a3aa50323c6b0facff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-07 14:19:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"5b43e4a38dd85d9114a5794570f67679abf42327","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2015-06-07 14:22:12.000000000","message":"Patch Set 6: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":6},{"id":"5a9797412e95252b836092925070aa9e20f5cdba","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2015-06-07 14:23:15.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)\n\n-1 to avoid accidental merge, I did a +2 +A just before and realized about something.","accounts_in_message":[],"_revision_number":6},{"id":"b876218be2b5981da4cbc29f6bdadba02d5ec970","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-07 14:25:01.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/6/gate/gate-puppet-keystone-puppet-lint/b16cac1/ : SUCCESS in 2m 17s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/6/gate/gate-puppet-keystone-puppet-syntax/8f42302/ : SUCCESS in 10s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/6/gate/gate-puppet-keystone-puppet-unit-3.3/2e464b4/ : SUCCESS in 3m 45s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/6/gate/gate-puppet-keystone-puppet-unit-3.4/a87f179/ : SUCCESS in 3m 22s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/6/gate/gate-puppet-keystone-puppet-unit-3.6/4c0d0d4/ : SUCCESS in 3m 41s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/6/gate/gate-puppet-keystone-puppet-unit-3.7/138b7c3/ : SUCCESS in 3m 19s\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/6/gate/gate-puppet-keystone-puppet-syntax-future/58b7b74/ : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":6},{"id":"afab5a3d6049ab82511eb742b322d4521e34ef94","author":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"date":"2015-06-11 02:44:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"98dbb12ab4249f87b0c3b582bece19f31d96a15c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-11 02:52:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-lint/89d19bc/ : SUCCESS in 3m 19s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-syntax/036953e/ : SUCCESS in 8s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-unit-3.3/f169af9/ : SUCCESS in 4m 35s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-unit-3.4/7282941/ : SUCCESS in 4m 45s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-unit-3.6/ba98393/ : SUCCESS in 5m 33s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-unit-3.7/bca959a/ : SUCCESS in 4m 13s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-unit-4.0/fb12bde/ : SUCCESS in 3m 11s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/4c17120/ : SUCCESS in 5m 53s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/8570b87/ : SUCCESS in 4m 52s (non-voting)\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/7/check/gate-puppet-keystone-puppet-syntax-future/f75fe9f/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":7},{"id":"b39646a729c579b9176c427bd9b0bf676f6b5ca7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-11 04:08:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (check-tripleo pipeline).\n\n- check-tripleo-ironic-overcloud-f21puppet-nonha http://logs.openstack.org/41/185441/7/check-tripleo/check-tripleo-ironic-overcloud-f21puppet-nonha/223225a/ : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":7},{"id":"6c0f7dd792e08968bd42e1c9624c9d710d15214a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2015-06-11 04:32:45.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks for this work!","accounts_in_message":[],"_revision_number":7},{"id":"86b5127bb4f62795be8516e760a9cd64b62fffdc","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2015-06-11 08:51:03.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5692139695d7d390255bd78568b8ca81d9226b04","author":{"_account_id":10540,"name":"Clayton O\u0027Neill","email":"clayton.oneill@charter.com","username":"coneill"},"date":"2015-06-11 17:38:38.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"d2f7ae4b0ec41b07ba2659dac1b29d9c4d9dfb9f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-11 17:39:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"b07eba7da9a92b3e29dffc766d1c6b103eee3a81","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-11 17:45:48.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/41/185441/7/gate/gate-puppet-keystone-puppet-lint/2615e97/ : SUCCESS in 3m 21s\n- gate-puppet-keystone-puppet-syntax http://logs.openstack.org/41/185441/7/gate/gate-puppet-keystone-puppet-syntax/693f8f0/ : SUCCESS in 12s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/41/185441/7/gate/gate-puppet-keystone-puppet-unit-3.3/6fc2706/ : SUCCESS in 4m 32s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/41/185441/7/gate/gate-puppet-keystone-puppet-unit-3.4/4d981bf/ : SUCCESS in 4m 49s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/41/185441/7/gate/gate-puppet-keystone-puppet-unit-3.6/1f8dc40/ : SUCCESS in 4m 04s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/41/185441/7/gate/gate-puppet-keystone-puppet-unit-3.7/4574db7/ : SUCCESS in 4m 57s\n- gate-puppet-keystone-puppet-syntax-future http://logs.openstack.org/41/185441/7/gate/gate-puppet-keystone-puppet-syntax-future/26e697d/ : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":7},{"id":"f5a89ca0b80508934e28893e4b08a910f2e0c889","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-11 17:45:49.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"bca04a399e6de63bd6d3d3d985aeaf3134802ac1","revisions":{"d7e967706977b528176dddcc917635d47eb13c50":{"kind":"REWORK","_number":1,"created":"2015-05-25 18:48:32.000000000","uploader":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"ref":"refs/changes/41/185441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/41/185441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/1"}}},"commit":{"parents":[{"commit":"eb240c5aa120d428567fb7c239167def96b6b75f","subject":"Merge \"Decouple sync_db from enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/eb240c5aa120d428567fb7c239167def96b6b75f"}]}],"author":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-25 18:47:09.000000000","tz":-360},"committer":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-25 18:47:49.000000000","tz":-360},"subject":"Add support for Fernet Tokens","message":"Add support for Fernet Tokens\n\nThis adds support for Fernet token creation during the installation and\nconfig options for Fernet tokens.\n\nChange-Id: I43657e9ba46f07c316a9e67a134e30aa3a44a867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d7e967706977b528176dddcc917635d47eb13c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d7e967706977b528176dddcc917635d47eb13c50"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb240c5aa120d428567fb7c239167def96b6b75f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d49f38df6b91dc3acf6949fd035e89a647f14a97":{"kind":"REWORK","_number":2,"created":"2015-05-25 19:03:41.000000000","uploader":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"ref":"refs/changes/41/185441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/41/185441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/2"}}},"commit":{"parents":[{"commit":"eb240c5aa120d428567fb7c239167def96b6b75f","subject":"Merge \"Decouple sync_db from enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/eb240c5aa120d428567fb7c239167def96b6b75f"}]}],"author":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-25 18:47:09.000000000","tz":-360},"committer":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-25 19:03:37.000000000","tz":-360},"subject":"Add support for Fernet Tokens","message":"Add support for Fernet Tokens\n\nThis adds support for Fernet token creation during the installation and\nconfig options for Fernet tokens.\n\nChange-Id: I43657e9ba46f07c316a9e67a134e30aa3a44a867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d49f38df6b91dc3acf6949fd035e89a647f14a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d49f38df6b91dc3acf6949fd035e89a647f14a97"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb240c5aa120d428567fb7c239167def96b6b75f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d0711f5e7edb55c50ecb101189e4300b911aa6ba":{"kind":"REWORK","_number":3,"created":"2015-05-26 03:13:30.000000000","uploader":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"ref":"refs/changes/41/185441/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/41/185441/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/3"}}},"commit":{"parents":[{"commit":"eb240c5aa120d428567fb7c239167def96b6b75f","subject":"Merge \"Decouple sync_db from enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/eb240c5aa120d428567fb7c239167def96b6b75f"}]}],"author":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-25 18:47:09.000000000","tz":-360},"committer":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-26 03:13:24.000000000","tz":-360},"subject":"Add support for Fernet Tokens","message":"Add support for Fernet Tokens\n\nThis adds support for Fernet token creation during the installation and\nconfig options for Fernet tokens.\n\nChange-Id: I43657e9ba46f07c316a9e67a134e30aa3a44a867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d0711f5e7edb55c50ecb101189e4300b911aa6ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d0711f5e7edb55c50ecb101189e4300b911aa6ba"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb240c5aa120d428567fb7c239167def96b6b75f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5359cd832a3b253b9e620b377b25da58a18f0bd0":{"kind":"REWORK","_number":4,"created":"2015-05-26 19:37:58.000000000","uploader":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"ref":"refs/changes/41/185441/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/41/185441/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/4"}}},"commit":{"parents":[{"commit":"eb240c5aa120d428567fb7c239167def96b6b75f","subject":"Merge \"Decouple sync_db from enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/eb240c5aa120d428567fb7c239167def96b6b75f"}]}],"author":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-25 18:47:09.000000000","tz":-360},"committer":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-26 19:37:51.000000000","tz":-360},"subject":"Add support for Fernet Tokens","message":"Add support for Fernet Tokens\n\nThis adds support for Fernet token creation during the installation and\nconfig options for Fernet tokens.\n\nChange-Id: I43657e9ba46f07c316a9e67a134e30aa3a44a867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5359cd832a3b253b9e620b377b25da58a18f0bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5359cd832a3b253b9e620b377b25da58a18f0bd0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb240c5aa120d428567fb7c239167def96b6b75f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"14d6fc979318b9bbe7aa5b639ad60648381bfac3":{"kind":"REWORK","_number":5,"created":"2015-05-26 20:27:22.000000000","uploader":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"ref":"refs/changes/41/185441/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/41/185441/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/5"}}},"commit":{"parents":[{"commit":"eb240c5aa120d428567fb7c239167def96b6b75f","subject":"Merge \"Decouple sync_db from enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/eb240c5aa120d428567fb7c239167def96b6b75f"}]}],"author":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-25 18:47:09.000000000","tz":-360},"committer":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-26 20:27:16.000000000","tz":-360},"subject":"Add support for Fernet Tokens","message":"Add support for Fernet Tokens\n\nThis adds support for Fernet token creation during the installation and\nconfig options for Fernet tokens.\n\nChange-Id: I43657e9ba46f07c316a9e67a134e30aa3a44a867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/14d6fc979318b9bbe7aa5b639ad60648381bfac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/14d6fc979318b9bbe7aa5b639ad60648381bfac3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb240c5aa120d428567fb7c239167def96b6b75f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"960142ed7660deb324f6a0e329e0c5e85a9d20c7":{"kind":"REWORK","_number":6,"created":"2015-05-29 03:59:15.000000000","uploader":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"ref":"refs/changes/41/185441/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/41/185441/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/6"}}},"commit":{"parents":[{"commit":"eb240c5aa120d428567fb7c239167def96b6b75f","subject":"Merge \"Decouple sync_db from enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/eb240c5aa120d428567fb7c239167def96b6b75f"}]}],"author":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-25 18:47:09.000000000","tz":-360},"committer":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-29 03:59:09.000000000","tz":-360},"subject":"Add support for Fernet Tokens","message":"Add support for Fernet Tokens\n\nThis adds support for Fernet token creation during the installation and\nconfig options for Fernet tokens.\n\nChange-Id: I43657e9ba46f07c316a9e67a134e30aa3a44a867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/960142ed7660deb324f6a0e329e0c5e85a9d20c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/960142ed7660deb324f6a0e329e0c5e85a9d20c7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb240c5aa120d428567fb7c239167def96b6b75f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bca04a399e6de63bd6d3d3d985aeaf3134802ac1":{"kind":"REWORK","_number":7,"created":"2015-06-11 02:44:50.000000000","uploader":{"_account_id":9500,"name":"Matt Fischer","email":"matt@mattfischer.com","username":"mfisch"},"ref":"refs/changes/41/185441/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/41/185441/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/41/185441/7"}}},"commit":{"parents":[{"commit":"58cdeac78a1bbeb98bc2450a1baf2edb6af150e1","subject":"Merge \"Restructures authentication for resource providers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/58cdeac78a1bbeb98bc2450a1baf2edb6af150e1"}]}],"author":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-05-25 18:47:09.000000000","tz":-360},"committer":{"name":"Matt Fischer","email":"matt@mattfischer.com","date":"2015-06-11 02:44:44.000000000","tz":-360},"subject":"Add support for Fernet Tokens","message":"Add support for Fernet Tokens\n\nThis adds support for Fernet token creation during the installation and\nconfig options for Fernet tokens.\n\nChange-Id: I43657e9ba46f07c316a9e67a134e30aa3a44a867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/bca04a399e6de63bd6d3d3d985aeaf3134802ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/bca04a399e6de63bd6d3d3d985aeaf3134802ac1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"58cdeac78a1bbeb98bc2450a1baf2edb6af150e1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
