)]}'
{"id":"openstack%2Fpuppet-keystone~208019","triplet_id":"openstack%2Fpuppet-keystone~stable%2Fkilo~I85226d862b63e227867ed43bc35d7fefc9e424cd","project":"openstack/puppet-keystone","branch":"stable/kilo","topic":"puppet/backports/kilo","hashtags":[],"change_id":"I85226d862b63e227867ed43bc35d7fefc9e424cd","subject":"WSGI: use real service name in restart_keystone Exec","status":"MERGED","created":"2015-07-31 18:48:26.000000000","updated":"2015-08-04 12:51:00.000000000","submitted":"2015-08-04 12:50:59.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6ce1331913588f7ed7d2a038ef25cbb0de37c15e","_number":208019,"virtual_id_number":208019,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"date":"2015-07-31 23:55:56.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":0,"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"value":2,"date":"2015-08-04 12:50:59.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2015-08-04 12:14:09.000000000","_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":2,"date":"2015-08-04 08:36:43.000000000","_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-08-03 15:49:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":1,"date":"2015-08-03 19:43:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2015-08-04 12:14:09.000000000","_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"value":0,"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-08-03 15:49:12.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2015-08-03 19:43:34.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2015-08-04 08:36:43.000000000","updated_by":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"reviewer":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"state":"REVIEWER"},{"updated":"2015-08-04 12:14:09.000000000","updated_by":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"reviewer":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"state":"REVIEWER"},{"updated":"2015-08-04 12:50:59.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"0bdffd91ddc0e71d5dddeeef47f69f8a156da534","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2015-07-31 18:49:01.000000000","message":"Topic set to puppet/backports/kilo","accounts_in_message":[],"_revision_number":1},{"id":"2251eb7ddbed6f654cddde16c2c0a7df8ee61b7f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-31 20:59:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-lint/8536c23/ : SUCCESS in 2m 15s\n- gate-puppet-keystone-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-syntax-3-dsvm-centos7/5dd57ba/ : NOT_REGISTERED\n- gate-puppet-keystone-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-syntax-4-dsvm-centos7/e8e91ee/ : NOT_REGISTERED\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.3/7a6079d/ : SUCCESS in 3m 26s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.4/cd06fe9/ : SUCCESS in 3m 37s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.6/7dc85a2/ : SUCCESS in 4m 00s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.7/58864c9/ : SUCCESS in 5m 34s\n- gate-puppet-keystone-puppet-unit-3.8 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.8/713e137/ : SUCCESS in 2m 27s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-4.0/7415efb/ : SUCCESS in 3m 16s\n- gate-puppet-keystone-puppet-unit-latest http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-latest/4101bd7/ : SUCCESS in 3m 23s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/506b69f/ : SUCCESS in 9m 23s\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/f608ca4/ : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bfa7ad203fc455173432d4b8dcee35c5a553ef12","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2015-07-31 23:55:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d39d0d5931a9132de6d1059d2de589082ae72a2c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-01 00:30:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-lint/c69d674/ : SUCCESS in 2m 44s\n- gate-puppet-keystone-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-syntax-3-dsvm-centos7/7f580ad/ : SUCCESS in 2m 37s\n- gate-puppet-keystone-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-syntax-4-dsvm-centos7/b713cb6/ : SUCCESS in 2m 28s\n- gate-puppet-keystone-puppet-unit-3.3 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.3/cc324f3/ : SUCCESS in 4m 34s\n- gate-puppet-keystone-puppet-unit-3.4 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.4/f41f61e/ : SUCCESS in 4m 47s\n- gate-puppet-keystone-puppet-unit-3.6 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.6/1ede46b/ : SUCCESS in 4m 58s\n- gate-puppet-keystone-puppet-unit-3.7 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.7/8ec43d6/ : SUCCESS in 5m 21s\n- gate-puppet-keystone-puppet-unit-3.8 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-3.8/b92c411/ : SUCCESS in 5m 01s\n- gate-puppet-keystone-puppet-unit-4.0 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-4.0/f2430e4/ : SUCCESS in 4m 11s\n- gate-puppet-keystone-puppet-unit-latest http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-unit-latest/acc0fa0/ : SUCCESS in 4m 30s (non-voting)\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/5abe9c1/ : SUCCESS in 9m 06s\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/19/208019/1/check/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/4b3556f/ : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e21fa6fcb8a9c4cefe0482ff4ee52e856d36da5b","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2015-08-03 15:49:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7c0d5974b21f4780dd2b06996aef566ab622fe9b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2015-08-03 19:43:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"340f166305b2cd94169b5f47a522c633403e9ed7","author":{"_account_id":7155,"name":"Sebastien Badia","email":"seb@sebian.fr","username":"sbadia"},"date":"2015-08-04 08:36:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb33077f8666d742dbfaae126935cb143701788b","author":{"_account_id":9410,"name":"Yanis Guenane","email":"yguenane@redhat.com","username":"yguenane"},"date":"2015-08-04 12:14:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbc4b4287121f05efbb32013b4ac343b6f6b4a08","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-04 12:14:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"1f8e07261fa4f378ad83b6fe1f6c7ac82c149d1c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-04 12:50:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-lint/7192856/ : SUCCESS in 3m 05s\n- gate-puppet-keystone-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-syntax-3-dsvm-centos7/e7a5173/ : SUCCESS in 4m 24s\n- gate-puppet-keystone-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-syntax-4-dsvm-centos7/32f1e3a/ : SUCCESS in 3m 26s\n- gate-puppet-keystone-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-unit-3.3-dsvm-centos7/80ed8c4/ : SUCCESS in 4m 11s\n- gate-puppet-keystone-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-unit-3.4-dsvm-centos7/136b699/ : SUCCESS in 4m 57s\n- gate-puppet-keystone-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-unit-3.6-dsvm-centos7/ae5893d/ : SUCCESS in 4m 58s\n- gate-puppet-keystone-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-unit-3.7-dsvm-centos7/5c9ce01/ : SUCCESS in 5m 12s\n- gate-puppet-keystone-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-unit-3.8-dsvm-centos7/e26dfec/ : SUCCESS in 5m 12s\n- gate-puppet-keystone-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-unit-4.0-dsvm-centos7/a7e1ee6/ : SUCCESS in 4m 22s\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-beaker-rspec-dsvm-trusty/359a1ff/ : SUCCESS in 10m 49s\n- gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/19/208019/1/gate/gate-puppet-keystone-puppet-beaker-rspec-dsvm-centos7/23fcdc8/ : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6ce1331913588f7ed7d2a038ef25cbb0de37c15e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-04 12:51:00.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb19dec488ee0e9e72d368189c4cba7da17cbb37","revisions":{"fb19dec488ee0e9e72d368189c4cba7da17cbb37":{"kind":"REWORK","_number":1,"created":"2015-07-31 18:48:26.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/19/208019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/19/208019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/19/208019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/19/208019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/19/208019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/19/208019/1"}}},"commit":{"parents":[{"commit":"70b17397f929f199105cc1adad87f92fb8668472","subject":"Fix rspec 3.x syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/70b17397f929f199105cc1adad87f92fb8668472"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2015-07-30 14:16:30.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2015-07-31 18:48:26.000000000","tz":0},"subject":"WSGI: use real service name in restart_keystone Exec","message":"WSGI: use real service name in restart_keystone Exec\n\nBefore, when running the Exec which restart Keystone service, we used\n$service_name but this is wrong because service_name can be keystone,\nopenstack-keystone or httpd.\nOn Ubuntu/Debian, Apache service name is not httpd but apache2.\n\nSo the service could never be restarted if we were 1/ running\nUbuntu/Debian 2/ changing the default domain name.\n\nThis patch looks up in puppetlabs-apache the real Apache service name\nwhen $service_name is httpd (if WSGI is enabled and eventlet disabled).\n\nThat way, we are sure the service name that will be used will the right\none.\nAlso add some unit tests that validate this change.\n\nChange-Id: I85226d862b63e227867ed43bc35d7fefc9e424cd\nCloses-bug: #1479783\n(cherry picked from commit f654d3cd9b8ce57345823e544047b311856c7fa7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fb19dec488ee0e9e72d368189c4cba7da17cbb37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fb19dec488ee0e9e72d368189c4cba7da17cbb37"}]},"parents_data":[{"branch_name":"refs/heads/stable/kilo","commit_id":"70b17397f929f199105cc1adad87f92fb8668472","is_merged_in_target_branch":true,"change_id":"I84b993065a00f5b5c55bd6834c6d54bc57c249e0","change_number":207445,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/kilo"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
