)]}'
{"id":"openstack%2Fpuppet-keystone~325533","triplet_id":"openstack%2Fpuppet-keystone~master~I69b94f2fa47e9f3fd153ef5097fe306070a5758e","project":"openstack/puppet-keystone","branch":"master","topic":"fix_wsgi","hashtags":[],"change_id":"I69b94f2fa47e9f3fd153ef5097fe306070a5758e","subject":"Fix incorrect parameter","status":"MERGED","created":"2016-06-04 06:19:07.000000000","updated":"2016-06-06 11:00:43.000000000","submitted":"2016-06-06 11:00:43.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c6c58e08b764c46853d225cd48e8b5e77159d682","_number":325533,"virtual_id_number":325533,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":2,"date":"2016-06-06 11:00:43.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"value":0,"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"all":[{"value":2,"date":"2016-06-06 10:23:11.000000000","_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-06-06 09:44:24.000000000","_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"value":1,"date":"2016-06-06 10:08:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"all":[{"value":1,"date":"2016-06-06 10:23:11.000000000","_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"value":0,"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-06 09:44:24.000000000","updated_by":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"reviewer":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"state":"REVIEWER"},{"updated":"2016-06-06 10:08:53.000000000","updated_by":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"reviewer":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"state":"REVIEWER"},{"updated":"2016-06-06 10:23:11.000000000","updated_by":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"reviewer":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"state":"REVIEWER"},{"updated":"2016-06-06 11:00:43.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"7cf2f15533a66c0877fab76565965707947f2ef4","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2016-06-04 06:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48398288331a64b9f0f2085068f32e49d8913087","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-04 07:04:35.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/33/325533/1/check-tripleo/gate-tripleo-ci-f22-nonha/bc34ea8/ : FAILURE in 45m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0e042deab63767b85e12e63ca24a34f60886dbf3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-04 07:24:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-trusty/778ebf8/ : SUCCESS in 18m 52s\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/69044f6/ : SUCCESS in 20m 29s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-lint/4f1331b/ : SUCCESS in 3m 49s\n- gate-puppet-keystone-puppet-syntax-3-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-syntax-3-centos-7/7eed009/ : SUCCESS in 2m 37s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-syntax-4-centos-7/a898cea/ : SUCCESS in 2m 33s\n- gate-puppet-keystone-puppet-unit-3.6-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-unit-3.6-centos-7/5094080/ : SUCCESS in 6m 21s\n- gate-puppet-keystone-puppet-unit-3.8-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-unit-3.8-centos-7/5b94dc9/ : SUCCESS in 7m 00s\n- gate-puppet-keystone-puppet-unit-4.5-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-unit-4.5-centos-7/0a0a431/ : SUCCESS in 6m 25s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-unit-latest-centos-7/3082888/ : SUCCESS in 6m 35s (non-voting)\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/60823ac/ : FAILURE in 1h 00m 22s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/4432797/ : SUCCESS in 33m 43s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/4f167b3/ : SUCCESS in 30m 47s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/ea348a3/ : SUCCESS in 25m 01s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial/f1fdc5b/ : SUCCESS in 29m 18s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial/e14d3b3/ : SUCCESS in 24m 15s","accounts_in_message":[],"_revision_number":1},{"id":"140d9bebe8668e8a1a6454166425fc6ce37ac274","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-06-04 07:40:38.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/2036/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/2023/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/2069/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"45917fea05ad431a121124585788ce20c38b359c","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2016-06-04 20:03:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7178324ebf028f6b8122fa804e10aa65f0a2aae7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-04 20:48:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-trusty/5e26766/ : SUCCESS in 16m 56s\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/1176e9d/ : SUCCESS in 20m 26s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-lint/8bc3eca/ : SUCCESS in 2m 49s\n- gate-puppet-keystone-puppet-syntax-3-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-syntax-3-centos-7/d52cc60/ : SUCCESS in 2m 38s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-syntax-4-centos-7/2b7efc3/ : SUCCESS in 2m 38s\n- gate-puppet-keystone-puppet-unit-3.6-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-unit-3.6-centos-7/8116347/ : SUCCESS in 5m 01s\n- gate-puppet-keystone-puppet-unit-3.8-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-unit-3.8-centos-7/8b52041/ : SUCCESS in 5m 49s\n- gate-puppet-keystone-puppet-unit-4.5-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-unit-4.5-centos-7/ecf6d9f/ : SUCCESS in 5m 26s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-keystone-puppet-unit-latest-centos-7/3921e9d/ : SUCCESS in 6m 17s (non-voting)\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/b427a88/ : SUCCESS in 36m 05s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/f4080ad/ : SUCCESS in 39m 28s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/014adf8/ : SUCCESS in 30m 35s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/95070a1/ : SUCCESS in 25m 37s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial/dcef716/ : SUCCESS in 29m 56s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/33/325533/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial/5d4731b/ : SUCCESS in 23m 55s","accounts_in_message":[],"_revision_number":1},{"id":"046cac728f05973f2204b17b02537d4e990b53eb","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-06-04 21:47:46.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.neutron_vlan_ha/2040/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/2027/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/2073/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"382456e7b5567301cc0314d983e6486e33bd2e77","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-04 22:50:17.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/33/325533/1/check-tripleo/gate-tripleo-ci-f22-nonha/b49714c/ : FAILURE in 2h 46m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b5b80daf255909d48422c2af79f6026712919457","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-06-06 09:44:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"21a169660504deaa6c1713ed0434edf20474c94d","author":{"_account_id":14007,"name":"Maksym Yatsenko","email":"myatsenko@mirantis.com","username":"myatsenko"},"date":"2016-06-06 10:08:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b751e413645c5a13f20b602b22e53e2206ec35ee","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-06-06 10:23:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"231b5d41203b26d86cf135cb5ef47cad847d5445","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 10:23:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"34d16a4bebdfb2dea14c3da04350299a696cdeb8","date":"2016-06-06 11:00:43.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"1a401a583b73f9e6b8ac3b59b342fc5f04c065e5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 11:00:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-trusty http://logs.openstack.org/33/325533/1/gate/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-trusty/2c84dfb/ : SUCCESS in 17m 19s\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/33/325533/1/gate/gate-puppet-keystone-puppet-beaker-rspec-centos-7/557909d/ : SUCCESS in 22m 02s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/33/325533/1/gate/gate-puppet-keystone-puppet-lint/c53d862/ : SUCCESS in 3m 40s\n- gate-puppet-keystone-puppet-syntax-3-centos-7 http://logs.openstack.org/33/325533/1/gate/gate-puppet-keystone-puppet-syntax-3-centos-7/84de62b/ : SUCCESS in 2m 48s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/33/325533/1/gate/gate-puppet-keystone-puppet-syntax-4-centos-7/eb7d809/ : SUCCESS in 3m 10s\n- gate-puppet-keystone-puppet-unit-3.6-centos-7 http://logs.openstack.org/33/325533/1/gate/gate-puppet-keystone-puppet-unit-3.6-centos-7/1177e61/ : SUCCESS in 5m 16s\n- gate-puppet-keystone-puppet-unit-3.8-centos-7 http://logs.openstack.org/33/325533/1/gate/gate-puppet-keystone-puppet-unit-3.8-centos-7/bee5586/ : SUCCESS in 6m 40s\n- gate-puppet-keystone-puppet-unit-4.5-centos-7 http://logs.openstack.org/33/325533/1/gate/gate-puppet-keystone-puppet-unit-4.5-centos-7/345035f/ : SUCCESS in 6m 58s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/33/325533/1/gate/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/4a052f5/ : SUCCESS in 34m 07s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/33/325533/1/gate/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/866626d/ : SUCCESS in 35m 32s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/33/325533/1/gate/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/6006f26/ : SUCCESS in 29m 34s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/33/325533/1/gate/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/3e9c48b/ : SUCCESS in 29m 02s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/33/325533/1/gate/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial/7265a6e/ : SUCCESS in 29m 09s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/33/325533/1/gate/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial/7c58623/ : SUCCESS in 23m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e0cc49ea6d16f12939106131ce56db784a08f01","revisions":{"7e0cc49ea6d16f12939106131ce56db784a08f01":{"kind":"REWORK","_number":1,"created":"2016-06-04 06:19:07.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/33/325533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/33/325533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/33/325533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/33/325533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/33/325533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/33/325533/1"}}},"commit":{"parents":[{"commit":"6cbdce249d7d00f7922a3735bf89897274ec1f87","subject":"Merge \"Prepare 9.0.0 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/6cbdce249d7d00f7922a3735bf89897274ec1f87"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2016-06-04 06:12:02.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2016-06-04 06:18:56.000000000","tz":-180},"subject":"Fix incorrect parameter","message":"Fix incorrect parameter\n\nplatform_params should be platform_parameters\n\nChange-Id: I69b94f2fa47e9f3fd153ef5097fe306070a5758e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7e0cc49ea6d16f12939106131ce56db784a08f01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7e0cc49ea6d16f12939106131ce56db784a08f01"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6cbdce249d7d00f7922a3735bf89897274ec1f87","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
