)]}'
{"id":"openstack%2Fpuppet-keystone~334791","triplet_id":"openstack%2Fpuppet-keystone~master~Id934a56e047baefe6f0c9a26c512f43fbdc793ae","project":"openstack/puppet-keystone","branch":"master","hashtags":[],"change_id":"Id934a56e047baefe6f0c9a26c512f43fbdc793ae","subject":"Support member_role_id and member_role_name config options","status":"MERGED","created":"2016-06-28 06:36:40.000000000","updated":"2016-06-28 14:38:31.000000000","submitted":"2016-06-28 14:38:30.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"817166c2a866df6a2a37e1537dee0b1101738eff","_number":334791,"virtual_id_number":334791,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-06-28 14:38:30.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-06-28 09:37:52.000000000","_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"value":2,"date":"2016-06-28 13:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"value":1,"date":"2016-06-28 13:55:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-28 09:37:52.000000000","updated_by":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"reviewer":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"state":"REVIEWER"},{"updated":"2016-06-28 13:55:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2016-06-28 14:38:30.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ff9438e26e3432774c64b3b4ba78933b1184753d","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2016-06-28 06:36:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56fcf2305d60c5f39175a0bc90fcb7f5c6171222","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-28 07:17:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/1c6ef38/ : SUCCESS in 24m 02s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/14cf693/ : SUCCESS in 17m 49s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-lint/8fc47d9/ : SUCCESS in 3m 15s\n- gate-puppet-keystone-puppet-syntax-3-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-syntax-3-centos-7/a1a9bc5/ : SUCCESS in 3m 05s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-syntax-4-centos-7/b0b7583/ : SUCCESS in 2m 58s\n- gate-puppet-keystone-puppet-unit-3.6-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-3.6-centos-7/95d9f90/ : SUCCESS in 7m 52s\n- gate-puppet-keystone-puppet-unit-3.8-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-3.8-centos-7/e8f49ab/ : SUCCESS in 7m 41s\n- gate-puppet-keystone-puppet-unit-4.5-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-4.5-centos-7/92f8d0a/ : SUCCESS in 12m 06s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-latest-centos-7/8985195/ : SUCCESS in 9m 23s (non-voting)\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/fab58b1/ : SUCCESS in 36m 55s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/91e85d1/ : FAILURE in 20m 32s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/58a93cb/ : SUCCESS in 33m 50s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial-nv http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial-nv/0c624b1/ : SUCCESS in 32m 40s (non-voting)\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial-nv http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial-nv/406bbfb/ : SUCCESS in 28m 39s (non-voting)\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial-nv http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial-nv/916c952/ : SUCCESS in 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fbcea3cbaf3cd58edeac5a76f337fb7683cf1138","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-28 08:13:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-nonha http://logs.openstack.org/91/334791/1/check-tripleo/gate-tripleo-ci-centos-7-nonha/e3ac4d2/ : SUCCESS in 1h 36m 25s","accounts_in_message":[],"_revision_number":1},{"id":"58cb0775eb3cdc7b415489cd0ce4fe5fc9e7ed3c","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-06-28 08:23:15.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/2801/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/2858/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/714/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"17811ab4111469ad7ca5b42585ecede01c43f0eb","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-06-28 09:37:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8a6a1ad8077e30603b63f0ce171814ff5859b3a9","author":{"_account_id":7732,"name":"IvanBerezovskiy","email":"ivan.berezovskii@gmail.com","username":"iberezovskiy"},"date":"2016-06-28 09:37:52.000000000","message":"Patch Set 1: Code-Review+2\n\nInit class becomes bigger and bigger :(","accounts_in_message":[],"_revision_number":1},{"id":"15bfcb0baccc7d3dd2a335f877a7f028f77bec1c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-28 10:16:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/5566cff/ : SUCCESS in 20m 59s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/2e99a20/ : SUCCESS in 18m 40s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-lint/ff35893/ : SUCCESS in 4m 34s\n- gate-puppet-keystone-puppet-syntax-3-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-syntax-3-centos-7/681b6a5/ : SUCCESS in 3m 26s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-syntax-4-centos-7/8fcfac6/ : SUCCESS in 6m 11s\n- gate-puppet-keystone-puppet-unit-3.6-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-3.6-centos-7/c130ec4/ : SUCCESS in 8m 42s\n- gate-puppet-keystone-puppet-unit-3.8-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-3.8-centos-7/56f9242/ : SUCCESS in 8m 20s\n- gate-puppet-keystone-puppet-unit-4.5-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-4.5-centos-7/32216aa/ : SUCCESS in 6m 14s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-latest-centos-7/0e7fb18/ : SUCCESS in 7m 59s (non-voting)\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/3ba3bbe/ : SUCCESS in 26m 10s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/67ec219/ : FAILURE in 15m 13s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/4cb231d/ : SUCCESS in 34m 39s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial-nv http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial-nv/b5410f7/ : SUCCESS in 32m 40s (non-voting)\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial-nv http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial-nv/4fda0b1/ : SUCCESS in 35m 23s (non-voting)\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial-nv http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial-nv/9ff7db9/ : SUCCESS in 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed4592b7e63c1c20ba77f4240cf835f1dfa9e380","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-06-28 11:17:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/2807/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/2863/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/719/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"76d7f4e6902a0dc9e631828f8eb2597e5bb199f4","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2016-06-28 11:34:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"222ece6c1b07dd5df41498d484cf82963f9a2be0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-28 11:48:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-nonha http://logs.openstack.org/91/334791/1/check-tripleo/gate-tripleo-ci-centos-7-nonha/15e7b6b/ : SUCCESS in 2h 06m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3196d85d7821948f4daa7971ca0df3971fb1da6d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-28 12:18:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/e4abc67/ : SUCCESS in 27m 03s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/5dedff4/ : SUCCESS in 20m 43s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-lint/3b54bfb/ : SUCCESS in 4m 03s\n- gate-puppet-keystone-puppet-syntax-3-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-syntax-3-centos-7/59c62da/ : SUCCESS in 4m 55s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-syntax-4-centos-7/254d2fe/ : SUCCESS in 5m 40s\n- gate-puppet-keystone-puppet-unit-3.6-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-3.6-centos-7/be797a6/ : SUCCESS in 8m 09s\n- gate-puppet-keystone-puppet-unit-3.8-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-3.8-centos-7/0c237d4/ : SUCCESS in 8m 05s\n- gate-puppet-keystone-puppet-unit-4.5-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-4.5-centos-7/1d01a58/ : SUCCESS in 8m 09s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-keystone-puppet-unit-latest-centos-7/cde2d41/ : SUCCESS in 8m 24s (non-voting)\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/300796e/ : SUCCESS in 38m 04s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/2e502d8/ : SUCCESS in 37m 20s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/eb9d0b7/ : SUCCESS in 32m 33s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial-nv http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial-nv/8073976/ : SUCCESS in 25m 15s (non-voting)\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial-nv http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial-nv/a8cc10c/ : SUCCESS in 34m 30s (non-voting)\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial-nv http://logs.openstack.org/91/334791/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial-nv/1b81054/ : SUCCESS in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5308106b4bd24ce91bdf2ef151c9bf9b151a1e7f","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-06-28 13:08:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/2811/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/2867/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/723/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b3df4450b439064e200d89dbafa3904cf436cdaf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-06-28 13:55:12.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68c5c198313dd17d4fe0066905f63e33841c4f15","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-28 13:55:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"1f4a45fea9e974fc177143b6cc2b1791898e7bcd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-28 14:38:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/334791/1/gate/gate-puppet-keystone-puppet-beaker-rspec-centos-7/cf0ecf2/ : SUCCESS in 24m 06s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/91/334791/1/gate/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/6bf6bd0/ : SUCCESS in 20m 21s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/91/334791/1/gate/gate-puppet-keystone-puppet-lint/2ee0e75/ : SUCCESS in 4m 46s\n- gate-puppet-keystone-puppet-syntax-3-centos-7 http://logs.openstack.org/91/334791/1/gate/gate-puppet-keystone-puppet-syntax-3-centos-7/b74f52a/ : SUCCESS in 3m 41s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/91/334791/1/gate/gate-puppet-keystone-puppet-syntax-4-centos-7/0f9c21a/ : SUCCESS in 5m 22s\n- gate-puppet-keystone-puppet-unit-3.6-centos-7 http://logs.openstack.org/91/334791/1/gate/gate-puppet-keystone-puppet-unit-3.6-centos-7/d55b65f/ : SUCCESS in 10m 16s\n- gate-puppet-keystone-puppet-unit-3.8-centos-7 http://logs.openstack.org/91/334791/1/gate/gate-puppet-keystone-puppet-unit-3.8-centos-7/1dd9ac6/ : SUCCESS in 7m 34s\n- gate-puppet-keystone-puppet-unit-4.5-centos-7 http://logs.openstack.org/91/334791/1/gate/gate-puppet-keystone-puppet-unit-4.5-centos-7/ebc8cd9/ : SUCCESS in 9m 48s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/91/334791/1/gate/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/d0b4aa6/ : SUCCESS in 27m 10s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/91/334791/1/gate/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/51f6e85/ : SUCCESS in 35m 24s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/91/334791/1/gate/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/6b25d00/ : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":1},{"id":"817166c2a866df6a2a37e1537dee0b1101738eff","date":"2016-06-28 14:38:31.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51c68fb127c28b9748b352871783c3ec6ef5c83b","revisions":{"51c68fb127c28b9748b352871783c3ec6ef5c83b":{"kind":"REWORK","_number":1,"created":"2016-06-28 06:36:40.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/91/334791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/91/334791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/91/334791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/91/334791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/91/334791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/91/334791/1"}}},"commit":{"parents":[{"commit":"7c34d6c6ff4417f14b1d882ddfcdbfa118044be4","subject":"Merge \"Fix idempotency of keystone_tenant description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7c34d6c6ff4417f14b1d882ddfcdbfa118044be4"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2016-06-28 06:36:15.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2016-06-28 06:36:15.000000000","tz":600},"subject":"Support member_role_id and member_role_name config options","message":"Support member_role_id and member_role_name config options\n\nChange-Id: Id934a56e047baefe6f0c9a26c512f43fbdc793ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/51c68fb127c28b9748b352871783c3ec6ef5c83b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/51c68fb127c28b9748b352871783c3ec6ef5c83b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7c34d6c6ff4417f14b1d882ddfcdbfa118044be4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
