)]}'
{"id":"openstack%2Fpuppet-keystone~492202","triplet_id":"openstack%2Fpuppet-keystone~master~I13a0af479dcac45ff77685f5eccfb865f7dab0f5","project":"openstack/puppet-keystone","branch":"master","topic":"bug/1709519","hashtags":[],"change_id":"I13a0af479dcac45ff77685f5eccfb865f7dab0f5","subject":"fix ldap packages for keystone backend","status":"MERGED","created":"2017-08-09 16:54:53.000000000","updated":"2017-08-11 11:20:07.000000000","submitted":"2017-08-11 10:21:10.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"892a3db5bc43db488f61eec27f2a1ae57549568e","_number":492202,"virtual_id_number":492202,"owner":{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-08-11 10:21:10.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-08-10 01:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2017-08-09 21:12:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-08-10 01:05:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2017-08-11 03:46:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-09 17:08:09.000000000","updated_by":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-08-09 20:52:23.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-08-09 20:53:41.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2017-08-10 01:05:03.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2017-08-11 03:46:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2017-08-11 10:21:10.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5556bc546a830d1ed2480c6626e866557d11daef","author":{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},"date":"2017-08-09 16:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"017055ab19a00857d9c61a7c0e98ae4457b25a74","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2017-08-09 17:08:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/8541/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/10722/ : FAILURE\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/10605/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"30341c038a385c181881b997f1f10a2d4e44a6ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-09 20:25:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/02/492202/1/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/35364c3/ : SUCCESS in 34m 05s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/02/492202/1/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/efb4314/ : SUCCESS in 29m 50s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/02/492202/1/check/gate-puppet-keystone-puppet-lint/944948a/ : SUCCESS in 2m 42s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/02/492202/1/check/gate-puppet-keystone-puppet-syntax-4-centos-7/5b38a45/ : SUCCESS in 3m 19s\n- gate-puppet-keystone-puppet-unit-4.8-centos-7 http://logs.openstack.org/02/492202/1/check/gate-puppet-keystone-puppet-unit-4.8-centos-7/ee80eaa/ : SUCCESS in 12m 43s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/02/492202/1/check/gate-puppet-keystone-puppet-unit-latest-centos-7/2b2dcc7/ : SUCCESS in 11m 12s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/02/492202/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/bc1f521/ : SUCCESS in 51m 17s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/02/492202/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/28cd5dd/ : SUCCESS in 52m 01s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/02/492202/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/0806c96/ : SUCCESS in 51m 43s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/02/492202/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/911c29d/ : SUCCESS in 36m 26s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/7891cc6/ : SUCCESS in 43m 29s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/58c8b13/ : SUCCESS in 45m 17s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/f3621e7/ : SUCCESS in 34m 15s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/7d19bf2/ : SUCCESS in 27m 04s\n- gate-puppet-keystone-releasenotes http://docs-draft.openstack.org/02/492202/1/check/gate-puppet-keystone-releasenotes/5c339df//releasenotes/build/html/ : SUCCESS in 2m 03s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/02/492202/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/8c549d3/ : SUCCESS in 1h 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d7b80716770a3c0d8330e4201efa80e244ac2424","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-08-09 21:03:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ca8757b67b7febf9ef2a1d926c0ff1acd8cd66d","author":{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},"date":"2017-08-09 21:07:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3966505cc34ace3d0d2505fba93ae2acbc3e709","author":{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},"date":"2017-08-09 21:07:49.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nfixed release note issue","accounts_in_message":[],"_revision_number":1},{"id":"7d5c9cdcba2cb2384d13210adbb3e86c6fefd45b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-08-09 21:12:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f9c0a12515c5b09e40d7927d785b60ead65e4a7f","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2017-08-09 21:26:30.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/8545/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/10726/ : FAILURE\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/10609/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"4fa5a31316974b8a177906a490e3117a7a6bcd56","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-09 23:55:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/9844263/ : FAILURE in 18m 29s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/6c11542/ : SUCCESS in 31m 35s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-lint/30948de/ : SUCCESS in 7m 24s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-syntax-4-centos-7/43d7cd0/ : SUCCESS in 18m 00s\n- gate-puppet-keystone-puppet-unit-4.8-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-unit-4.8-centos-7/20353d2/ : SUCCESS in 56m 58s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-unit-latest-centos-7/b9a0a76/ : SUCCESS in 47m 23s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/038ff23/ : SUCCESS in 54m 38s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/ee4b058/ : SUCCESS in 56m 27s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/3b9bbcd/ : SUCCESS in 59m 52s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/98f71e4/ : SUCCESS in 40m 26s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/5c92215/ : SUCCESS in 54m 03s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/fda9c82/ : SUCCESS in 59m 04s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/92b225e/ : SUCCESS in 36m 22s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/91aadca/ : SUCCESS in 32m 41s\n- gate-puppet-keystone-releasenotes http://docs-draft.openstack.org/02/492202/2/check/gate-puppet-keystone-releasenotes/3ba26f5//releasenotes/build/html/ : SUCCESS in 1m 28s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/02/492202/2/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/5c09ba2/ : SUCCESS in 1h 47m 29s","accounts_in_message":[],"_revision_number":2},{"id":"4cd7fce2cfe516df8474ddfc0382b2ce1c57f9ea","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-08-10 01:05:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"261f7c659f45548a96f21217c01e53ee5aae9b15","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-08-10 01:05:10.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"82d63e2615174c6588ec55f2ff822bf64d7c64ad","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2017-08-10 01:24:36.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/8547/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/10728/ : FAILURE\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/10611/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d3423bb83dc675c673cc9a57b3ff492d8a24244c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-10 02:43:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/14eb5f2/ : SUCCESS in 36m 56s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/36089d2/ : SUCCESS in 28m 55s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-lint/f283516/ : SUCCESS in 3m 02s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-syntax-4-centos-7/e02e744/ : SUCCESS in 3m 22s\n- gate-puppet-keystone-puppet-unit-4.8-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-unit-4.8-centos-7/c5d80ba/ : SUCCESS in 13m 09s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-unit-latest-centos-7/75479b8/ : SUCCESS in 10m 52s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/c0d30c7/ : SUCCESS in 54m 04s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/c8dea2b/ : SUCCESS in 56m 33s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/61031ae/ : FAILURE in 1h 06m 01s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/4d9255f/ : FAILURE in 1h 06m 24s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/c369565/ : SUCCESS in 55m 46s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/2b99f72/ : SUCCESS in 46m 45s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/98da79f/ : SUCCESS in 35m 04s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/540720e/ : SUCCESS in 34m 51s\n- gate-puppet-keystone-releasenotes http://docs-draft.openstack.org/02/492202/2/check/gate-puppet-keystone-releasenotes/353aa5e//releasenotes/build/html/ : SUCCESS in 1m 37s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/02/492202/2/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/99397b9/ : SUCCESS in 1h 31m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b46cd6c41c92ae6653e741d660c894b7373b1006","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-08-10 03:45:30.000000000","message":"Patch Set 2:\n\nretry","accounts_in_message":[],"_revision_number":2},{"id":"4d9d7811d9eb49f9a7848d43a95ed255bf741dfe","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-08-10 03:45:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"03d6e141287a5e179f32624a02f6697ade29caae","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2017-08-10 04:05:29.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/8553/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/10734/ : FAILURE\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/10617/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3e94a1ac4e6e3fad6a62ac9422f01910a1e0ab16","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-10 05:32:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/096f1a9/ : SUCCESS in 36m 06s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/6fdf513/ : SUCCESS in 30m 21s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-lint/09fa585/ : SUCCESS in 2m 43s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-syntax-4-centos-7/b384ce0/ : SUCCESS in 4m 51s\n- gate-puppet-keystone-puppet-unit-4.8-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-unit-4.8-centos-7/07fc685/ : SUCCESS in 38m 52s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-unit-latest-centos-7/0e5f12e/ : SUCCESS in 11m 48s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/67a52af/ : SUCCESS in 55m 24s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/7e4cdeb/ : SUCCESS in 49m 12s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/92f2544/ : SUCCESS in 52m 28s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/1c73e55/ : SUCCESS in 43m 11s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/2765144/ : SUCCESS in 36m 24s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/5a68591/ : SUCCESS in 53m 23s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/5e91e65/ : SUCCESS in 32m 26s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/e40414b/ : SUCCESS in 28m 16s\n- gate-puppet-keystone-releasenotes http://docs-draft.openstack.org/02/492202/2/check/gate-puppet-keystone-releasenotes/43414ad//releasenotes/build/html/ : SUCCESS in 1m 13s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/02/492202/2/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/2a9eba4/ : SUCCESS in 1h 34m 15s","accounts_in_message":[],"_revision_number":2},{"id":"55a1bf6be5c85d875a1d7b560eadd3e4435f33b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-10 05:32:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"54703f61c8fb14dead9201a691ee0d38848dc365","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-08-10 14:01:30.000000000","message":"Abandoned\n\nNeed to clear out the queue as this is going to fail. Will restore momentarily","accounts_in_message":[],"_revision_number":2},{"id":"978a27b11d074a35bb2e1ceead4cbaf322b5b423","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-08-10 15:11:09.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"466ef7eb1779e3433f73b64e73a250f66bff471d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-08-10 16:08:26.000000000","message":"Patch Set 2: Workflow-1\n\ni\u0027ll lift my -1 once the gate is clear","accounts_in_message":[],"_revision_number":2},{"id":"b735a5d97a49dba432f61da7c34a2f05d0c0914c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-10 19:17:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-beaker-rspec-centos-7/8e4dd7a/ : SUCCESS in 40m 38s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/76c3d4f/ : SUCCESS in 31m 18s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-lint/29995b7/ : SUCCESS in 3m 19s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-syntax-4-centos-7/48657a4/ : SUCCESS in 4m 20s\n- gate-puppet-keystone-puppet-unit-4.8-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-unit-4.8-centos-7/53dd290/ : SUCCESS in 32m 56s\n- gate-puppet-keystone-puppet-unit-latest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-keystone-puppet-unit-latest-centos-7/7728f1f/ : SUCCESS in 14m 36s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/5d0caa3/ : SUCCESS in 57m 17s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/37ab10a/ : SUCCESS in 49m 35s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/9847bbf/ : SUCCESS in 52m 16s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/10adff3/ : SUCCESS in 35m 34s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/6b0d348/ : SUCCESS in 35m 30s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/5954593/ : SUCCESS in 50m 10s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/13f3ee8/ : SUCCESS in 31m 40s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/1b047f5/ : SUCCESS in 33m 03s\n- gate-puppet-keystone-releasenotes http://docs-draft.openstack.org/02/492202/2/check/gate-puppet-keystone-releasenotes/ee527a6//releasenotes/build/html/ : SUCCESS in 1m 32s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/02/492202/2/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/e1f9beb/ : SUCCESS in 1h 32m 50s","accounts_in_message":[],"_revision_number":2},{"id":"9e1ee71e41a58c24061cd72703317fc1580ddd0d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-08-11 03:46:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"01d1d772bd777eaa0ae9a5fb6a28239ad09f538b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-11 03:46:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"58f9da524a401cddc2b08ee0b097fefbb49137cf","date":"2017-08-11 10:21:10.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"c7cad4181a6a6c873c7f54aa171906ade787e3cc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-11 10:21:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-keystone-puppet-beaker-rspec-centos-7 http://logs.openstack.org/02/492202/2/gate/gate-puppet-keystone-puppet-beaker-rspec-centos-7/3bb0830/ : SUCCESS in 37m 40s\n- gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/02/492202/2/gate/gate-puppet-keystone-puppet-beaker-rspec-ubuntu-xenial/1601a48/ : SUCCESS in 33m 12s\n- gate-puppet-keystone-puppet-lint http://logs.openstack.org/02/492202/2/gate/gate-puppet-keystone-puppet-lint/8470265/ : SUCCESS in 3m 25s\n- gate-puppet-keystone-puppet-syntax-4-centos-7 http://logs.openstack.org/02/492202/2/gate/gate-puppet-keystone-puppet-syntax-4-centos-7/a30f12b/ : SUCCESS in 4m 14s\n- gate-puppet-keystone-puppet-unit-4.8-centos-7 http://logs.openstack.org/02/492202/2/gate/gate-puppet-keystone-puppet-unit-4.8-centos-7/321e51b/ : SUCCESS in 14m 09s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/02/492202/2/gate/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/8afc5c2/ : SUCCESS in 52m 10s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/02/492202/2/gate/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/69175cb/ : SUCCESS in 52m 14s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/02/492202/2/gate/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/dce3108/ : SUCCESS in 58m 17s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/02/492202/2/gate/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/99e69bc/ : SUCCESS in 42m 15s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/gate/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/d9b9c57/ : SUCCESS in 50m 38s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/gate/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/01e5f51/ : SUCCESS in 51m 51s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/gate/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/fa825bf/ : SUCCESS in 31m 21s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/02/492202/2/gate/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/2f6e0ef/ : SUCCESS in 44m 26s\n- gate-puppet-keystone-releasenotes http://docs-draft.openstack.org/02/492202/2/gate/gate-puppet-keystone-releasenotes/702839f//releasenotes/build/html/ : SUCCESS in 1m 35s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/02/492202/2/gate/gate-tripleo-ci-centos-7-nonha-multinode-oooq/7dbec63/ : SUCCESS in 1h 48m 22s","accounts_in_message":[],"_revision_number":2},{"id":"eaafbca9d17bb2232c5a0fa5359caf6abfce58f3","author":{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},"date":"2017-08-11 11:20:07.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ocata as commit 45577a94ab054f275305011b9a7895a78f13f619","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ba0d32d5b6d9012fe80ce32fdadf1b36d19c33e","revisions":{"a76c6ae7c8e6fbe3e0374a8307df5f76c8862dfc":{"kind":"REWORK","_number":1,"created":"2017-08-09 16:54:53.000000000","uploader":{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},"ref":"refs/changes/02/492202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/02/492202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/02/492202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/02/492202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/02/492202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/02/492202/1"}}},"commit":{"parents":[{"commit":"19e4cfffe093bc601d8f3df691b3d18efd5642f0","subject":"Merge \"Add vhost access/error logs file/syslog options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/19e4cfffe093bc601d8f3df691b3d18efd5642f0"}]}],"author":{"name":"Matthew J. Black","email":"mjblack@gmail.com","date":"2017-08-09 16:13:59.000000000","tz":-240},"committer":{"name":"Matthew J. Black","email":"mjblack@gmail.com","date":"2017-08-09 16:53:36.000000000","tz":-240},"subject":"fix ldap packages for keystone backend","message":"fix ldap packages for keystone backend\n\nWhen puppet runs it will try to install python-ldap and\npython-ldappool. Each run will install one or the other\ndue to the other package settings telling the package\nmanager to uninstall it.\n\nChange-Id: I13a0af479dcac45ff77685f5eccfb865f7dab0f5\nCloses-Bug: #1709519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/a76c6ae7c8e6fbe3e0374a8307df5f76c8862dfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/a76c6ae7c8e6fbe3e0374a8307df5f76c8862dfc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"19e4cfffe093bc601d8f3df691b3d18efd5642f0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8ba0d32d5b6d9012fe80ce32fdadf1b36d19c33e":{"kind":"REWORK","_number":2,"created":"2017-08-09 21:07:07.000000000","uploader":{"_account_id":8318,"name":"Matthew J Black","email":"mjblack@gmail.com","username":"mjblack"},"ref":"refs/changes/02/492202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/02/492202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/02/492202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/02/492202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/02/492202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/02/492202/2"}}},"commit":{"parents":[{"commit":"19e4cfffe093bc601d8f3df691b3d18efd5642f0","subject":"Merge \"Add vhost access/error logs file/syslog options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/19e4cfffe093bc601d8f3df691b3d18efd5642f0"}]}],"author":{"name":"Matthew J. Black","email":"mjblack@gmail.com","date":"2017-08-09 16:13:59.000000000","tz":-240},"committer":{"name":"Matthew J. Black","email":"mjblack@gmail.com","date":"2017-08-09 21:07:01.000000000","tz":-240},"subject":"fix ldap packages for keystone backend","message":"fix ldap packages for keystone backend\n\nWhen puppet runs it will try to install python-ldap and\npython-ldappool. Each run will install one or the other\ndue to the other package settings telling the package\nmanager to uninstall it.\n\nChange-Id: I13a0af479dcac45ff77685f5eccfb865f7dab0f5\nCloses-Bug: #1709519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8ba0d32d5b6d9012fe80ce32fdadf1b36d19c33e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8ba0d32d5b6d9012fe80ce32fdadf1b36d19c33e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"19e4cfffe093bc601d8f3df691b3d18efd5642f0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
