)]}'
{"id":"openstack%2Fpuppet-keystone~534129","triplet_id":"openstack%2Fpuppet-keystone~master~I60e9d31737e8134ba52162c977f74c66601d7f14","project":"openstack/puppet-keystone","branch":"master","topic":"fix_test_requirements","hashtags":[],"change_id":"I60e9d31737e8134ba52162c977f74c66601d7f14","subject":"Remove pinned versions from test-requirements.txt","status":"MERGED","created":"2018-01-16 06:46:44.000000000","updated":"2018-01-17 13:59:10.000000000","submitted":"2018-01-17 13:59:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"534129-1516197550932-cf179ea8","meta_rev_id":"a8e5229ac391a4de16b6b4bec4c12ea9336b68bc","_number":534129,"virtual_id_number":534129,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2018-01-17 13:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2018-01-16 18:16:52.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":-1,"date":"2018-01-16 22:40:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2018-01-17 13:54:23.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-16 22:40:29.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2018-01-17 13:54:23.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2018-01-17 13:59:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc5c3c520649cc40a9f7d8dfe5204b7448644424","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2018-01-16 06:46:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2f75a8340c3b8dc5a703330fd98be4bc430be79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-16 07:14:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes http://logs.openstack.org/29/534129/1/check/build-openstack-releasenotes/f5b5c3b/html/ : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cacf62af0b352f52f83676e46ebb60ca41255e19","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-01-16 18:16:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28aa667a91b65d6fdf70dc7e3eda260ae3f7e587","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2018-01-16 22:40:29.000000000","message":"Patch Set 1: Code-Review-1\n\nnot sure about it","accounts_in_message":[],"_revision_number":1},{"id":"b8702417d93b83183865004e87e931e39c69c960","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-01-17 13:54:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c1192fe3730efe8d1301829688355fe259cc39e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 13:54:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c25f54137304fecfec1a15b11c63ac1efe84762b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 13:59:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes http://logs.openstack.org/29/534129/1/gate/build-openstack-releasenotes/59ee99c/html/ : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f2ae6f3ef8ce1fc533901c05f956ad203ec7429b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 13:59:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2865cc7e6bac2e745c3db47b81e2aa515cd484a8","revisions":{"2865cc7e6bac2e745c3db47b81e2aa515cd484a8":{"kind":"REWORK","_number":1,"created":"2018-01-16 06:46:44.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/29/534129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/29/534129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/29/534129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/29/534129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/29/534129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/29/534129/1"}}},"commit":{"parents":[{"commit":"23b8e8045693473d2149a19924d3a9fe1abe03cd","subject":"Add group to policy management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/23b8e8045693473d2149a19924d3a9fe1abe03cd"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2018-01-16 04:29:45.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2018-01-16 04:29:45.000000000","tz":480},"subject":"Remove pinned versions from test-requirements.txt","message":"Remove pinned versions from test-requirements.txt\n\nAs we are using the upper constraints when running the release\nnotes task using tox, we do not need to set a specific version\nand we can rely on whatever is configured in upper constraints.\n\nChange-Id: I60e9d31737e8134ba52162c977f74c66601d7f14\nCo-Authored-By: Mohammed Naser \u003cmnaser@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/2865cc7e6bac2e745c3db47b81e2aa515cd484a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/2865cc7e6bac2e745c3db47b81e2aa515cd484a8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"23b8e8045693473d2149a19924d3a9fe1abe03cd","is_merged_in_target_branch":true,"change_id":"I4dfcf05aa8418df3ee1a13925f0831dc30921186","change_number":532665,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
