)]}'
{"id":"openstack%2Fpuppet-keystone~574115","triplet_id":"openstack%2Fpuppet-keystone~stable%2Fpike~I69783f75600048b9c4b2abb445b59e135029f3df","project":"openstack/puppet-keystone","branch":"stable/pike","topic":"bug/1557790-stable/pike","hashtags":[],"change_id":"I69783f75600048b9c4b2abb445b59e135029f3df","subject":"Use OS_BOOTSTRAP_PASSWORD instead of passing CLI","status":"MERGED","created":"2018-06-11 01:49:33.000000000","updated":"2018-06-12 09:05:48.000000000","submitted":"2018-06-12 09:05:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"574115-1528794348352-4330fb6e","meta_rev_id":"6d93c58d73bb33ec283ecaaae52850325ff26993","_number":574115,"virtual_id_number":574115,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2018-06-12 09:05:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2018-06-11 23:30:44.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2018-06-12 07:43:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-11 23:30:44.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2018-06-12 07:43:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2018-06-12 09:05:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d74ca8a7490584564c7495ad65586ea2aa0a2bf9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2018-06-11 01:49:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f84c1018e65d4ddcf83bed57015c77918e44af36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-11 03:07:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/15/574115/1/check/puppet-openstack-lint/56dbd06/ : SUCCESS in 3m 38s\n- puppet-openstack-syntax-4 http://logs.openstack.org/15/574115/1/check/puppet-openstack-syntax-4/9fe482c/ : SUCCESS in 3m 55s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/15/574115/1/check/puppet-openstack-unit-4.8-centos-7/4ae5f9b/ : SUCCESS in 16m 04s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/15/574115/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/8547480/ : SUCCESS in 1h 11m 04s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/15/574115/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/13cdcf0/ : SUCCESS in 1h 08m 15s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/15/574115/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/8421dc8/ : SUCCESS in 1h 07m 46s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/15/574115/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/8d0bd1a/ : SUCCESS in 47m 56s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/15/574115/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/4aef20d/ : SUCCESS in 35m 48s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/15/574115/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/148cf7b/ : SUCCESS in 36m 41s\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/15/574115/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/c58baa2/ : SUCCESS in 30m 08s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/15/574115/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/7dfeb7c/ : SUCCESS in 32m 30s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/15/574115/1/check/puppet-openstack-beaker-centos-7/584ca67/ : SUCCESS in 39m 16s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/15/574115/1/check/puppet-openstack-beaker-xenial/8342eba/ : SUCCESS in 29m 26s","accounts_in_message":[],"_revision_number":1},{"id":"905b841aae0416b32b440b228615b5da760d625b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-06-11 23:30:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28cd1e6346fd40c13df3dec0e36dd26771107520","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-06-12 07:43:36.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3129adc33908a25954a92764b84fc0d5d15238d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-12 07:43:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ac501d58d69c862288c3ca30a4fba58ed7d8e63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-12 09:05:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7427ba9e0771e8ae5309cef07c447e57b2745907","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-12 09:05:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/15/574115/1/gate/puppet-openstack-lint/b7a7e56/ : SUCCESS in 4m 52s\n- puppet-openstack-syntax-4 http://logs.openstack.org/15/574115/1/gate/puppet-openstack-syntax-4/d58b71d/ : SUCCESS in 5m 24s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/15/574115/1/gate/puppet-openstack-unit-4.8-centos-7/56c425d/ : SUCCESS in 17m 13s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/15/574115/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7/b6910d8/ : SUCCESS in 1h 10m 26s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/15/574115/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/bc00450/ : SUCCESS in 1h 02m 57s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/15/574115/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/c50b9f4/ : SUCCESS in 1h 07m 04s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/15/574115/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7/4665719/ : SUCCESS in 47m 17s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/15/574115/1/gate/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/c26bb4b/ : SUCCESS in 37m 07s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/15/574115/1/gate/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/c909c4f/ : SUCCESS in 37m 22s\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/15/574115/1/gate/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/94c6936/ : SUCCESS in 31m 26s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/15/574115/1/gate/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/0831fe5/ : SUCCESS in 30m 30s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/15/574115/1/gate/puppet-openstack-beaker-centos-7/0ba363b/ : SUCCESS in 44m 12s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/15/574115/1/gate/puppet-openstack-beaker-xenial/606586b/ : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bbf75097295c73b37e99a39a8d9b4d1c0629e61","revisions":{"1bbf75097295c73b37e99a39a8d9b4d1c0629e61":{"kind":"REWORK","_number":1,"created":"2018-06-11 01:49:33.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/15/574115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/15/574115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/15/574115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/15/574115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/15/574115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/15/574115/1"}}},"commit":{"parents":[{"commit":"4de23ac5cefb38b63524d170f9eb7adf0bbeba3a","subject":"Configure memcache_secret_key as secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/4de23ac5cefb38b63524d170f9eb7adf0bbeba3a"}]}],"author":{"name":"Tobias Urdin","email":"tobias@busybox.se","date":"2018-04-05 12:15:39.000000000","tz":120},"committer":{"name":"zhongshengping","email":"chdzsp@163.com","date":"2018-06-11 01:49:33.000000000","tz":0},"subject":"Use OS_BOOTSTRAP_PASSWORD instead of passing CLI","message":"Use OS_BOOTSTRAP_PASSWORD instead of passing CLI\n\nWhen bootstrapping use environment variable\nOS_BOOTSTRAP_PASSWORD instead of passing password\nover CLI since it\u0027s a possible leak.\n\nChange-Id: I69783f75600048b9c4b2abb445b59e135029f3df\nPartial-Bug: 1557790\n(cherry picked from commit b933d0bd1a69628f263d59894dd5f217415d45da)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1bbf75097295c73b37e99a39a8d9b4d1c0629e61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1bbf75097295c73b37e99a39a8d9b4d1c0629e61"}]},"parents_data":[{"branch_name":"refs/heads/stable/pike","commit_id":"4de23ac5cefb38b63524d170f9eb7adf0bbeba3a","is_merged_in_target_branch":true,"change_id":"I74afaf85569b4ba756f4ca82f822e481f80253b6","change_number":554163,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
