)]}'
{"id":"openstack%2Fpuppet-keystone~609308","triplet_id":"openstack%2Fpuppet-keystone~master~Ia5e3818f110114338afa978af12f5b85f3e2b861","project":"openstack/puppet-keystone","branch":"master","topic":"standalone-puppet-gates","hashtags":[],"change_id":"Ia5e3818f110114338afa978af12f5b85f3e2b861","subject":"Run tripleo standalone job","status":"ABANDONED","created":"2018-10-10 08:11:01.000000000","updated":"2018-10-19 15:30:24.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a6e70986c3513199339e2c7b4e4bfabecd15431","_number":609308,"virtual_id_number":609308,"owner":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2018-10-15 17:25:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2018-10-15 15:02:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2018-10-15 15:08:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-10 08:33:20.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2018-10-15 15:02:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-10-15 15:08:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2018-10-15 17:25:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae903b2e91865f1424535fd0ed40cde462dbc84a","author":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"date":"2018-10-10 08:11:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cb8c5a9a942ae3f5de437721710616c6104a575","author":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"date":"2018-10-10 08:21:43.000000000","message":"Topic set to standalone-puppet-gates","accounts_in_message":[],"_revision_number":1},{"id":"9fc808bd7bfa0e098975c48af1209e3ad0f49ce3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-10-10 08:29:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario-py3-tempest-centos-7 http://logs.rdoproject.org/08/609308/1/openstack-check/puppet-openstack-integration-5-scenario-py3-tempest-centos-7/1c0e9fe/ : SUCCESS in 14m 03s (non-voting)\n- puppet-openstack-integration-5-scenario-py3-tempest-fedora-28 http://logs.rdoproject.org/08/609308/1/openstack-check/puppet-openstack-integration-5-scenario-py3-tempest-fedora-28/cb6a118/ : SUCCESS in 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a345955f4ecab9be7abc19daa30a470efd409161","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2018-10-10 08:33:20.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f3527eb97aed9b1b3612de187864d96319382fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 10:26:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/08/609308/1/check/puppet-openstack-lint/cb11687/ : SUCCESS in 4m 53s\n- puppet-openstack-syntax-4 http://logs.openstack.org/08/609308/1/check/puppet-openstack-syntax-4/a79c9a4/ : SUCCESS in 4m 32s\n- puppet-openstack-syntax-5 http://logs.openstack.org/08/609308/1/check/puppet-openstack-syntax-5/e8dbaf7/ : SUCCESS in 3m 53s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/08/609308/1/check/puppet-openstack-unit-4.8-centos-7/ea8690e/ : SUCCESS in 15m 27s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/08/609308/1/check/puppet-openstack-unit-5.5-centos-7/9454d53/ : SUCCESS in 14m 31s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/08/609308/1/check/puppet-openstack-unit-latest-centos-7/bac9c37/ : SUCCESS in 14m 27s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/75a5889/ : SUCCESS in 57m 30s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/405313a/ : SUCCESS in 1h 07m 11s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/2777c4d/ : FAILURE in 1h 00m 52s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/27ba715/ : SUCCESS in 1h 03m 14s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/7234883/ : SUCCESS in 47m 17s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/2ba5d05/ : SUCCESS in 44m 36s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/dfc9b9f/ : SUCCESS in 46m 45s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/ae86bf7/ : SUCCESS in 54m 32s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/28aa087/ : SUCCESS in 35m 57s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/e72558c/ : SUCCESS in 57m 30s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/13eb77a/ : FAILURE in 41m 53s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/fb23fe6/ : FAILURE in 47m 38s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/08/609308/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/604f35b/ : SUCCESS in 42m 22s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/08/609308/1/check/puppet-openstack-beaker-centos-7/6f44f18/ : SUCCESS in 41m 21s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/08/609308/1/check/puppet-openstack-beaker-ubuntu-bionic/11501b4/ : SUCCESS in 33m 57s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/08/609308/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/fddddb6/ : SUCCESS in 1h 44m 24s\n- tripleo-puppet-ci-centos-7-standalone http://logs.openstack.org/08/609308/1/check/tripleo-puppet-ci-centos-7-standalone/d52f099/ : FAILURE in 38m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9d30b35b28454c813a4d34ca5872f8f59b2e1684","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-10-10 14:27:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df46e261e0cb8d66caf6527c289f1faae8a69e1d","author":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"date":"2018-10-15 14:23:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"258e8868a0ee5f54794cef1826baf0dd252f5b88","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-10-15 15:02:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario-py3-tempest-centos-7 http://logs.rdoproject.org/08/609308/2/openstack-check/puppet-openstack-integration-5-scenario-py3-tempest-centos-7/7d99d48/ : SUCCESS in 36m 39s (non-voting)\n- puppet-openstack-integration-5-scenario-py3-tempest-fedora-28 http://logs.rdoproject.org/08/609308/2/openstack-check/puppet-openstack-integration-5-scenario-py3-tempest-fedora-28/9a34123/ : FAILURE in 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b61feceb1bd4883041a7e863d0a524a3c47a0a84","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-10-15 15:08:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a5ef169b000ab75da53b3d7e76c588d299e08089","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-15 17:25:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/08/609308/2/check/puppet-openstack-lint/1b1c083/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-4 http://logs.openstack.org/08/609308/2/check/puppet-openstack-syntax-4/3b74c8c/ : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5 http://logs.openstack.org/08/609308/2/check/puppet-openstack-syntax-5/3f5c3ce/ : SUCCESS in 3m 52s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/08/609308/2/check/puppet-openstack-unit-4.8-centos-7/e7761a8/ : SUCCESS in 16m 42s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/08/609308/2/check/puppet-openstack-unit-5.5-centos-7/7809f2c/ : SUCCESS in 13m 34s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/08/609308/2/check/puppet-openstack-unit-latest-centos-7/7cafe5b/ : SUCCESS in 13m 46s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/e5c4e4a/ : SUCCESS in 52m 40s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/368daea/ : SUCCESS in 1h 01m 44s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/497d9af/ : SUCCESS in 1h 00m 37s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/5bf72f3/ : SUCCESS in 58m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/76ef1e1/ : SUCCESS in 49m 01s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/85318b1/ : SUCCESS in 46m 03s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/0a97a8a/ : SUCCESS in 1h 08m 25s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/8d19cf8/ : FAILURE in 52m 14s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/ea5f9f7/ : SUCCESS in 42m 05s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/a9ad5a4/ : SUCCESS in 1h 13m 23s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/6a9d708/ : FAILURE in 36m 58s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/4a15ea5/ : FAILURE in 52m 59s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/08/609308/2/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/2fe5309/ : SUCCESS in 47m 07s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/08/609308/2/check/puppet-openstack-beaker-centos-7/48de665/ : SUCCESS in 38m 43s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/08/609308/2/check/puppet-openstack-beaker-ubuntu-bionic/56b4d6d/ : SUCCESS in 31m 37s\n- tripleo-puppet-ci-centos-7-standalone http://logs.openstack.org/08/609308/2/check/tripleo-puppet-ci-centos-7-standalone/c1983f0/ : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":2},{"id":"2a6e70986c3513199339e2c7b4e4bfabecd15431","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-10-19 15:30:24.000000000","message":"Abandoned\n\nhttps://review.openstack.org/#/c/609938/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8b256f7e587f9c81c7896c2c4f49bcaaea83f6c","revisions":{"b81ff918853aa019ba2da24b0c2da6fdfacd9033":{"kind":"REWORK","_number":1,"created":"2018-10-10 08:11:01.000000000","uploader":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"ref":"refs/changes/08/609308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/08/609308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/08/609308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/08/609308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/08/609308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/08/609308/1"}}},"commit":{"parents":[{"commit":"c948359a1b4661b5162e81813f37b793f5e3d04e","subject":"Dissuade .gitignore references to personal tools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/c948359a1b4661b5162e81813f37b793f5e3d04e"}]}],"author":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-10-10 08:06:01.000000000","tz":120},"committer":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-10-10 08:10:47.000000000","tz":120},"subject":"Run tripleo standalone job","message":"Run tripleo standalone job\n\nTo cover issues like https://bugs.launchpad.net/tripleo/+bug/1797035\n\nDepends-On: https://review.openstack.org/#/c/609300/\nChange-Id: Ia5e3818f110114338afa978af12f5b85f3e2b861\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/b81ff918853aa019ba2da24b0c2da6fdfacd9033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/b81ff918853aa019ba2da24b0c2da6fdfacd9033"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c948359a1b4661b5162e81813f37b793f5e3d04e","is_merged_in_target_branch":true,"change_id":"I8f6f7375b1adf7b6aeb056db35caff21d7558d01","change_number":608550,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e8b256f7e587f9c81c7896c2c4f49bcaaea83f6c":{"kind":"REWORK","_number":2,"created":"2018-10-15 14:23:36.000000000","uploader":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"ref":"refs/changes/08/609308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/08/609308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/08/609308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/08/609308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/08/609308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/08/609308/2"}}},"commit":{"parents":[{"commit":"c948359a1b4661b5162e81813f37b793f5e3d04e","subject":"Dissuade .gitignore references to personal tools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/c948359a1b4661b5162e81813f37b793f5e3d04e"}]}],"author":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-10-10 08:06:01.000000000","tz":120},"committer":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-10-15 14:23:26.000000000","tz":120},"subject":"Run tripleo standalone job","message":"Run tripleo standalone job\n\nTo cover issues like https://bugs.launchpad.net/tripleo/+bug/1797035\n\nDepends-On: https://review.openstack.org/#/c/609300/\nChange-Id: Ia5e3818f110114338afa978af12f5b85f3e2b861\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/e8b256f7e587f9c81c7896c2c4f49bcaaea83f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/e8b256f7e587f9c81c7896c2c4f49bcaaea83f6c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c948359a1b4661b5162e81813f37b793f5e3d04e","is_merged_in_target_branch":true,"change_id":"I8f6f7375b1adf7b6aeb056db35caff21d7558d01","change_number":608550,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
