)]}'
{"id":"openstack%2Fpuppet-keystone~644965","triplet_id":"openstack%2Fpuppet-keystone~master~I284160e520fbb73a041c49ff1a3e1ef5f3caf59d","project":"openstack/puppet-keystone","branch":"master","topic":"feature/keystone/openidc","hashtags":[],"change_id":"I284160e520fbb73a041c49ff1a3e1ef5f3caf59d","subject":"enable oauth support in keystone::fededation::openidc","status":"MERGED","created":"2019-03-20 18:34:58.000000000","updated":"2019-03-25 16:57:40.000000000","submitted":"2019-03-25 16:57:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"644965-1553533060063-da6fe399","meta_rev_id":"bc90de75f0c89957476a04a6ca9864f51685c8b7","_number":644965,"virtual_id_number":644965,"owner":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-03-22 01:25:42.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":2,"date":"2019-03-25 16:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-22 02:09:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-03-21 17:53:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-03-25 14:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-03-25 14:55:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-20 18:35:12.000000000","updated_by":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"reviewer":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"state":"REVIEWER"},{"updated":"2019-03-21 17:53:55.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-03-22 01:25:42.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-03-22 02:09:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-25 14:55:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-03-25 16:57:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d350357a5d350aca14f71062c67301b4e88751ca","author":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"date":"2019-03-20 18:34:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd0c7b3d8e9a07f08dcd1f9f43c6fd63b958ea0c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-20 19:51:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/65/644965/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/076ca46/ : SUCCESS in 34m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/65/644965/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e05418b/ : SUCCESS in 38m 17s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/65/644965/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/32ff217/ : FAILURE in 1h 12m 56s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/65/644965/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/6b5d106/ : SUCCESS in 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a30edf633a56486e873cb44b77e63a7f2eaa36a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-20 20:26:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/65/644965/1/check/puppet-openstack-lint-ubuntu-bionic/b6b995f/ : FAILURE in 4m 24s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/65/644965/1/check/puppet-openstack-syntax-4-ubuntu-bionic/e9c10d0/ : SUCCESS in 4m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/65/644965/1/check/puppet-openstack-syntax-5-ubuntu-bionic/4d4b4be/ : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/65/644965/1/check/puppet-openstack-syntax-6-ubuntu-bionic/67dbe85/ : SUCCESS in 4m 35s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/65/644965/1/check/puppet-openstack-unit-4.8-centos-7/0e3cd8f/ : SUCCESS in 17m 15s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/65/644965/1/check/puppet-openstack-unit-5.5-centos-7/9d19503/ : SUCCESS in 14m 44s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/65/644965/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/c4a88f7/ : SUCCESS in 10m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/65/644965/1/check/puppet-openstack-unit-latest-ubuntu-bionic/cf40fd6/ : FAILURE in 10m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/65/644965/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/911b01b/ : SUCCESS in 53m 31s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/65/644965/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/725621e/ : SUCCESS in 57m 38s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/65/644965/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/86104f5/ : SUCCESS in 1h 10m 08s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/65/644965/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/8217c40/ : SUCCESS in 1h 04m 02s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/0c68184/ : FAILURE in 20m 05s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/c69ccda/ : FAILURE in 31m 28s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/4b21144/ : FAILURE in 16m 11s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/a965a91/ : FAILURE in 27m 01s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/65/644965/1/check/puppet-openstack-beaker-centos-7/f3e133a/ : SUCCESS in 37m 06s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/65/644965/1/check/puppet-openstack-beaker-ubuntu-bionic/cbd7912/ : FAILURE in 32m 46s (non-voting)\n- puppet-keystone-tripleo-standalone http://logs.openstack.org/65/644965/1/check/puppet-keystone-tripleo-standalone/8c5d55d/ : SUCCESS in 1h 42m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9bfb8d67273b8ff7b90b0a92401601246c764dd2","author":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"date":"2019-03-20 20:27:11.000000000","message":"Topic changed from feature/openidc to feature/keystone/openidc","accounts_in_message":[],"_revision_number":1},{"id":"6a75de0731b876d40a1716b54624d50e93e7dcaa","author":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"date":"2019-03-20 20:28:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d32bfeb1c2df6a161e4577bc4c9043e56b979c95","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-20 21:24:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/65/644965/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/6e86dc7/ : SUCCESS in 40m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/65/644965/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e755f98/ : SUCCESS in 53m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/65/644965/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/da720fa/ : SUCCESS in 35m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/65/644965/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/736c1ab/ : FAILURE in 29m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8a81aaa4e2c2cf9e3185e7a38901c693516f212","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-20 22:12:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/65/644965/2/check/puppet-openstack-lint-ubuntu-bionic/e240608/ : SUCCESS in 4m 29s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/65/644965/2/check/puppet-openstack-syntax-4-ubuntu-bionic/672ebd1/ : SUCCESS in 7m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/65/644965/2/check/puppet-openstack-syntax-5-ubuntu-bionic/a2f0c8a/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/65/644965/2/check/puppet-openstack-syntax-6-ubuntu-bionic/73fa1c4/ : SUCCESS in 6m 51s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/65/644965/2/check/puppet-openstack-unit-4.8-centos-7/cfd1839/ : FAILURE in 15m 50s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/65/644965/2/check/puppet-openstack-unit-5.5-centos-7/1a52119/ : FAILURE in 15m 16s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/65/644965/2/check/puppet-openstack-unit-6.0-ubuntu-bionic/2a0b411/ : FAILURE in 10m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/65/644965/2/check/puppet-openstack-unit-latest-ubuntu-bionic/809cf42/ : FAILURE in 10m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/65/644965/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/6e1636a/ : SUCCESS in 1h 06m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/65/644965/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/ab065af/ : SUCCESS in 58m 58s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/65/644965/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/bd4c689/ : FAILURE in 1h 13m 12s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/65/644965/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/313f8fd/ : SUCCESS in 43m 59s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/da1765e/ : FAILURE in 32m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/7001839/ : FAILURE in 21m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/ac3806c/ : FAILURE in 24m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/7382d5a/ : FAILURE in 23m 02s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/65/644965/2/check/puppet-openstack-beaker-centos-7/d07e0f6/ : SUCCESS in 42m 33s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/65/644965/2/check/puppet-openstack-beaker-ubuntu-bionic/d3458ee/ : FAILURE in 26m 54s (non-voting)\n- puppet-keystone-tripleo-standalone http://logs.openstack.org/65/644965/2/check/puppet-keystone-tripleo-standalone/d4730b7/ : SUCCESS in 1h 38m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d62599a3b1a497535d72a40a0b14c37d111a6c44","author":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"date":"2019-03-21 13:47:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e335a987e0529127985d60645043c6459cf5ce75","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-21 14:46:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/65/644965/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/43468b6/ : SUCCESS in 55m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/65/644965/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/a5f44f0/ : SUCCESS in 45m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/65/644965/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/ce58bf4/ : SUCCESS in 41m 56s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/65/644965/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/728db3e/ : FAILURE in 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"96525ca725908f66678c83f8fd0471b9565b08b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 15:30:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/65/644965/3/check/puppet-openstack-lint-ubuntu-bionic/dc1da35/ : SUCCESS in 4m 51s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/65/644965/3/check/puppet-openstack-syntax-4-ubuntu-bionic/cbe544a/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/65/644965/3/check/puppet-openstack-syntax-5-ubuntu-bionic/3f96dc7/ : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/65/644965/3/check/puppet-openstack-syntax-6-ubuntu-bionic/3e84c6d/ : SUCCESS in 4m 11s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/65/644965/3/check/puppet-openstack-unit-4.8-centos-7/6d74eb4/ : SUCCESS in 15m 31s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/65/644965/3/check/puppet-openstack-unit-5.5-centos-7/2172d86/ : SUCCESS in 17m 03s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/65/644965/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/85d1e5f/ : SUCCESS in 11m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/65/644965/3/check/puppet-openstack-unit-latest-ubuntu-bionic/31667cf/ : FAILURE in 10m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/65/644965/3/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/8cbe55c/ : SUCCESS in 53m 46s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/65/644965/3/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/157e1d1/ : SUCCESS in 1h 04m 04s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/65/644965/3/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/18cd4df/ : SUCCESS in 56m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/65/644965/3/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/054098a/ : SUCCESS in 48m 00s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/3/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/7a9eed7/ : FAILURE in 19m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/3/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/2d09756/ : FAILURE in 22m 04s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/3/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/988ef84/ : FAILURE in 16m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/3/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/d4fa856/ : FAILURE in 25m 42s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/65/644965/3/check/puppet-openstack-beaker-centos-7/f3f3cab/ : SUCCESS in 38m 51s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/65/644965/3/check/puppet-openstack-beaker-ubuntu-bionic/2b7f48d/ : FAILURE in 25m 01s (non-voting)\n- puppet-keystone-tripleo-standalone http://logs.openstack.org/65/644965/3/check/puppet-keystone-tripleo-standalone/6f50fe0/ : SUCCESS in 1h 40m 36s","accounts_in_message":[],"_revision_number":3},{"id":"92bb9a7ecf173418734a84e7d826bafc730d6ce2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-03-21 17:53:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"0197e86e22d04ac860a0008c6e67dfd8269a2131","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-03-21 17:53:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dd37af6b7248be87e92023b57882c826d87527c4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-21 19:36:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/65/644965/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/7b8815a/ : SUCCESS in 35m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/65/644965/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/a215620/ : SUCCESS in 40m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/65/644965/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/4e20ca1/ : SUCCESS in 49m 37s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/65/644965/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/81c2684/ : FAILURE in 42m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"388b7a63e2e294e4077bcb41fe11c19e3f76f61d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 20:03:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-lint-ubuntu-bionic/b5d5061/ : SUCCESS in 7m 07s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-syntax-4-ubuntu-bionic/bc41b2c/ : SUCCESS in 7m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-syntax-5-ubuntu-bionic/de750d4/ : SUCCESS in 4m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-syntax-6-ubuntu-bionic/c9c7f4a/ : SUCCESS in 5m 21s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-unit-4.8-centos-7/6a74793/ : SUCCESS in 17m 39s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-unit-5.5-centos-7/e13e700/ : SUCCESS in 15m 28s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/ed4887b/ : SUCCESS in 11m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-unit-latest-ubuntu-bionic/e10105a/ : FAILURE in 10m 54s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/c9c40b4/ : SUCCESS in 55m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/298e518/ : SUCCESS in 1h 01m 35s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/3ae288e/ : SUCCESS in 1h 03m 13s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/2e46917/ : TIMED_OUT in 2h 00m 58s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/99b8e3e/ : FAILURE in 19m 35s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/8264b81/ : FAILURE in 20m 40s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/85da06e/ : FAILURE in 15m 25s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/5fc425e/ : FAILURE in 25m 29s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-beaker-centos-7/12f2f89/ : SUCCESS in 41m 07s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-beaker-ubuntu-bionic/9a4159f/ : FAILURE in 26m 55s (non-voting)\n- puppet-keystone-tripleo-standalone http://logs.openstack.org/65/644965/4/check/puppet-keystone-tripleo-standalone/5755c74/ : SUCCESS in 1h 37m 49s","accounts_in_message":[],"_revision_number":4},{"id":"638118b6cf0365b3753f133a7e4429167c865e48","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-03-22 01:25:42.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d175536ffb9d5a35222a3a06a66fd42f672de9a4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-22 02:09:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/65/644965/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/fd3846b/ : SUCCESS in 35m 19s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/65/644965/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/9e50834/ : SUCCESS in 40m 42s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/65/644965/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/044b875/ : SUCCESS in 39m 27s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/65/644965/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/45c38fb/ : SUCCESS in 37m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b6b6c307dffdf31248be84fb9cd8042bae481fbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 03:08:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-lint-ubuntu-bionic/9983732/ : SUCCESS in 7m 40s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-syntax-4-ubuntu-bionic/75aefa4/ : SUCCESS in 7m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-syntax-5-ubuntu-bionic/ae7edcd/ : SUCCESS in 7m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-syntax-6-ubuntu-bionic/ff012df/ : SUCCESS in 7m 52s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-unit-4.8-centos-7/7463591/ : SUCCESS in 17m 14s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-unit-5.5-centos-7/07994da/ : SUCCESS in 15m 53s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/eb9450d/ : SUCCESS in 14m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-unit-latest-ubuntu-bionic/ad95dfd/ : FAILURE in 10m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/1fc7d75/ : SUCCESS in 1h 01m 30s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/b51d8e7/ : SUCCESS in 1h 06m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/ae9fa54/ : SUCCESS in 1h 13m 34s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/d116bf3/ : SUCCESS in 48m 31s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/a44499c/ : FAILURE in 22m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/6222da3/ : FAILURE in 20m 09s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/cfde8fc/ : FAILURE in 16m 29s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/65/644965/4/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/3d6a30a/ : FAILURE in 26m 12s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/65/644965/4/check/puppet-openstack-beaker-centos-7/1d1c18e/ : SUCCESS in 40m 42s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/65/644965/4/check/puppet-openstack-beaker-ubuntu-bionic/4a11d4e/ : FAILURE in 29m 08s (non-voting)\n- puppet-keystone-tripleo-standalone http://logs.openstack.org/65/644965/4/check/puppet-keystone-tripleo-standalone/6d89c2c/ : SUCCESS in 1h 37m 01s","accounts_in_message":[],"_revision_number":4},{"id":"22c1e281621cdbd669db25c1bc0f0b810ccdc6eb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-25 14:55:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8a5c1a4fb5155675b525ea162b5b1af6ec556a99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 14:55:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f5c7a880f65da548e8d0c5dc23f7961d7af6b996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 16:57:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/65/644965/4/gate/puppet-openstack-lint-ubuntu-bionic/2abda53/ : SUCCESS in 5m 13s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/65/644965/4/gate/puppet-openstack-syntax-4-ubuntu-bionic/1e908b9/ : SUCCESS in 4m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/65/644965/4/gate/puppet-openstack-syntax-5-ubuntu-bionic/d657cff/ : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/65/644965/4/gate/puppet-openstack-syntax-6-ubuntu-bionic/28c89e0/ : SUCCESS in 4m 13s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/65/644965/4/gate/puppet-openstack-unit-4.8-centos-7/4bd3de6/ : SUCCESS in 18m 05s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/65/644965/4/gate/puppet-openstack-unit-5.5-centos-7/58fac3f/ : SUCCESS in 14m 26s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/65/644965/4/gate/puppet-openstack-unit-6.0-ubuntu-bionic/c581671/ : SUCCESS in 14m 26s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/65/644965/4/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/35e8c6d/ : SUCCESS in 54m 05s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/65/644965/4/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/ae48a84/ : SUCCESS in 1h 12m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/65/644965/4/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/a80f6e9/ : SUCCESS in 1h 17m 07s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/65/644965/4/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/2c987d8/ : SUCCESS in 43m 35s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/65/644965/4/gate/puppet-openstack-beaker-centos-7/8be2940/ : SUCCESS in 35m 38s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/65/644965/4/gate/puppet-openstack-beaker-ubuntu-bionic/919cd70/ : FAILURE in 31m 55s (non-voting)\n- puppet-keystone-tripleo-standalone http://logs.openstack.org/65/644965/4/gate/puppet-keystone-tripleo-standalone/382c1ea/ : SUCCESS in 1h 49m 55s","accounts_in_message":[],"_revision_number":4},{"id":"bc90de75f0c89957476a04a6ca9864f51685c8b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 16:57:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"717d40798aa5d34b2b8a2e37b88d69bf798cd0eb","revisions":{"f2e7e4f4a3f381ab2b65c0fda43bc3cc99ed4b0c":{"kind":"REWORK","_number":1,"created":"2019-03-20 18:34:58.000000000","uploader":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"ref":"refs/changes/65/644965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/65/644965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/1"}}},"commit":{"parents":[{"commit":"d29e86aaebbd25a30c544f2895d823a36384e00b","subject":"fix typo in templates/openidc.conf.erb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d29e86aaebbd25a30c544f2895d823a36384e00b"}]}],"author":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2019-03-20 18:24:31.000000000","tz":-240},"committer":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2019-03-20 18:25:12.000000000","tz":-240},"subject":"enable oauth support in keystone::fededation::openidc","message":"enable oauth support in keystone::fededation::openidc\n\nadd parameters for enabling OAuth support, which is required for the\ncommand line tools to interact with an openid identity provider.\n\nChange-Id: I284160e520fbb73a041c49ff1a3e1ef5f3caf59d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/f2e7e4f4a3f381ab2b65c0fda43bc3cc99ed4b0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/f2e7e4f4a3f381ab2b65c0fda43bc3cc99ed4b0c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d29e86aaebbd25a30c544f2895d823a36384e00b","is_merged_in_target_branch":false,"change_id":"Ic2cf0a4490fa68c424cf77c09c9c93fb282c66d6","change_number":644722,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"01e6d9f50c4fbd6e2f3f5297ecd227f7b06b65e1":{"kind":"REWORK","_number":2,"created":"2019-03-20 20:28:46.000000000","uploader":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"ref":"refs/changes/65/644965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/65/644965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/2"}}},"commit":{"parents":[{"commit":"d29e86aaebbd25a30c544f2895d823a36384e00b","subject":"fix typo in templates/openidc.conf.erb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d29e86aaebbd25a30c544f2895d823a36384e00b"}]}],"author":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2019-03-20 18:24:31.000000000","tz":-240},"committer":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2019-03-20 20:28:39.000000000","tz":-240},"subject":"enable oauth support in keystone::fededation::openidc","message":"enable oauth support in keystone::fededation::openidc\n\nadd parameters for enabling OAuth support, which is required for the\ncommand line tools to interact with an openid identity provider.\n\nChange-Id: I284160e520fbb73a041c49ff1a3e1ef5f3caf59d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/01e6d9f50c4fbd6e2f3f5297ecd227f7b06b65e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/01e6d9f50c4fbd6e2f3f5297ecd227f7b06b65e1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d29e86aaebbd25a30c544f2895d823a36384e00b","is_merged_in_target_branch":false,"change_id":"Ic2cf0a4490fa68c424cf77c09c9c93fb282c66d6","change_number":644722,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bccd04be81c1b861797729243e79370aa14e5556":{"kind":"REWORK","_number":3,"created":"2019-03-21 13:47:52.000000000","uploader":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"ref":"refs/changes/65/644965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/65/644965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/3"}}},"commit":{"parents":[{"commit":"d29e86aaebbd25a30c544f2895d823a36384e00b","subject":"fix typo in templates/openidc.conf.erb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d29e86aaebbd25a30c544f2895d823a36384e00b"}]}],"author":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2019-03-20 18:24:31.000000000","tz":-240},"committer":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2019-03-21 13:47:45.000000000","tz":-240},"subject":"enable oauth support in keystone::fededation::openidc","message":"enable oauth support in keystone::fededation::openidc\n\nadd parameters for enabling OAuth support, which is required for the\ncommand line tools to interact with an openid identity provider.\n\nChange-Id: I284160e520fbb73a041c49ff1a3e1ef5f3caf59d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/bccd04be81c1b861797729243e79370aa14e5556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/bccd04be81c1b861797729243e79370aa14e5556"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d29e86aaebbd25a30c544f2895d823a36384e00b","is_merged_in_target_branch":false,"change_id":"Ic2cf0a4490fa68c424cf77c09c9c93fb282c66d6","change_number":644722,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"717d40798aa5d34b2b8a2e37b88d69bf798cd0eb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-21 17:53:51.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/65/644965/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/65/644965/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/65/644965/4"}}},"commit":{"parents":[{"commit":"dcbd368d5c54d5c3627688b6b4cd586754341748","subject":"fix typo in templates/openidc.conf.erb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/dcbd368d5c54d5c3627688b6b4cd586754341748"}]}],"author":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2019-03-20 18:24:31.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-03-21 17:53:51.000000000","tz":0},"subject":"enable oauth support in keystone::fededation::openidc","message":"enable oauth support in keystone::fededation::openidc\n\nadd parameters for enabling OAuth support, which is required for the\ncommand line tools to interact with an openid identity provider.\n\nChange-Id: I284160e520fbb73a041c49ff1a3e1ef5f3caf59d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/717d40798aa5d34b2b8a2e37b88d69bf798cd0eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/717d40798aa5d34b2b8a2e37b88d69bf798cd0eb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dcbd368d5c54d5c3627688b6b4cd586754341748","is_merged_in_target_branch":true,"change_id":"Ic2cf0a4490fa68c424cf77c09c9c93fb282c66d6","change_number":644722,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
