)]}'
{"id":"openstack%2Fpuppet-keystone~668812","triplet_id":"openstack%2Fpuppet-keystone~master~If9fc5c0bb5d6216b8656ee1673e1812c543de305","project":"openstack/puppet-keystone","branch":"master","topic":"bug/1835161","hashtags":[],"change_id":"If9fc5c0bb5d6216b8656ee1673e1812c543de305","subject":"Manage oslo_middleware/max_request_body_size","status":"MERGED","created":"2019-07-03 07:41:52.000000000","updated":"2019-07-03 17:21:30.000000000","submitted":"2019-07-03 17:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668812-1562174490335-dfe5b5ea","meta_rev_id":"adf2706e7968b306000cd2471222316d231417ea","_number":668812,"virtual_id_number":668812,"owner":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-07-03 17:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-03 09:41:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"recommended":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"all":[{"value":1,"date":"2019-07-03 11:08:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":2,"date":"2019-07-03 15:48:38.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-03 12:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":1,"date":"2019-07-03 15:48:38.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 09:41:06.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 11:08:14.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2019-07-03 12:54:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-07-03 15:48:38.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-03 17:21:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"242b5a8b07be435089c9c55c5b2578d870e70fff","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-03 07:41:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f361b4925f0d35b1471601be210a3eaccf21dd4","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-03 08:05:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b440ced7b2a39b98fc5b0bb4a5f7fde92a8b3d61","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-03 08:52:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9ea16887dc3a7792e2baecaee2ce04f984976fa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-03 08:56:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/12/668812/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/a463acd/ : SUCCESS in 33m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/12/668812/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/76664d7/ : SUCCESS in 49m 00s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/12/668812/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/cf508f0/ : SUCCESS in 34m 59s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/12/668812/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/b6b9327/ : SUCCESS in 29m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94a97dbf4ca2183a96dc41de9aea7804235cb04d","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-03 08:59:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bc5c4b16b9e78baa15d973c4009f073bf1cb34bc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-03 09:41:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/12/668812/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/e07fbae/ : SUCCESS in 33m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/12/668812/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/3691f78/ : SUCCESS in 37m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/12/668812/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/db55404/ : SUCCESS in 39m 44s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/12/668812/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/4afd08f/ : SUCCESS in 36m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2a3b0732f5b002f2f8d272adc5e59c95dc395ac3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 10:37:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/12/668812/3/check/puppet-openstack-lint-ubuntu-bionic/414459f/ : SUCCESS in 5m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/12/668812/3/check/puppet-openstack-syntax-5-ubuntu-bionic/fd5687a/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/12/668812/3/check/puppet-openstack-syntax-6-ubuntu-bionic/1b238fc/ : SUCCESS in 6m 19s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/12/668812/3/check/puppet-openstack-unit-5.5-centos-7/ff2cf5a/ : SUCCESS in 17m 12s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/12/668812/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/1e84ec2/ : SUCCESS in 13m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/12/668812/3/check/puppet-openstack-unit-latest-ubuntu-bionic/b696d0a/ : FAILURE in 9m 46s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/12/668812/3/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/eeb9bac/ : SUCCESS in 52m 59s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/12/668812/3/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/4c99e7e/ : SUCCESS in 1h 02m 23s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/12/668812/3/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/28e5cff/ : SUCCESS in 1h 05m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/12/668812/3/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/daee4b2/ : SUCCESS in 46m 58s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/12/668812/3/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/43ca978/ : FAILURE in 36m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/12/668812/3/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/7ee8277/ : FAILURE in 32m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/12/668812/3/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/a6dc9b1/ : FAILURE in 9m 03s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/12/668812/3/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/4623839/ : FAILURE in 10m 51s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/12/668812/3/check/puppet-openstack-beaker-centos-7/5a6ec8e/ : SUCCESS in 50m 31s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/12/668812/3/check/puppet-openstack-beaker-ubuntu-bionic/c3361d7/ : FAILURE in 46m 31s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/12/668812/3/check/build-openstack-releasenotes/08e809f/html/ : SUCCESS in 3m 58s\n- puppet-keystone-tripleo-standalone http://logs.openstack.org/12/668812/3/check/puppet-keystone-tripleo-standalone/a3637cb/ : SUCCESS in 1h 33m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d559e887af4faf17aeffff04667c6168e9f9a1e2","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-03 11:08:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"81b58e71d59ee37b427339cede879b29684171a9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-07-03 12:54:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d0031036470877498cc973f5be2d7ff3b847b2ae","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-03 15:48:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"556ca1f2aaeae15f1a002446851532b9e0cd3b95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:48:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"77d14e7493d82167d604ee0b8cd9273e9c3835c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 17:21:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/12/668812/3/gate/puppet-openstack-lint-ubuntu-bionic/ad7e113/ : SUCCESS in 6m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/12/668812/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/3eb1371/ : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/12/668812/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/cb405d5/ : SUCCESS in 6m 10s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/12/668812/3/gate/puppet-openstack-unit-5.5-centos-7/352173c/ : SUCCESS in 17m 52s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/12/668812/3/gate/puppet-openstack-unit-6.4-ubuntu-bionic/dbd51b6/ : SUCCESS in 12m 25s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/12/668812/3/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/f2009cc/ : SUCCESS in 51m 51s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/12/668812/3/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/53c9d21/ : SUCCESS in 55m 36s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/12/668812/3/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/9d7a727/ : SUCCESS in 1h 06m 42s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/12/668812/3/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/ad97b70/ : SUCCESS in 51m 55s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/12/668812/3/gate/puppet-openstack-beaker-centos-7/88ba136/ : SUCCESS in 53m 29s\n- build-openstack-releasenotes http://logs.openstack.org/12/668812/3/gate/build-openstack-releasenotes/b107e56/html/ : SUCCESS in 3m 08s\n- puppet-keystone-tripleo-standalone http://logs.openstack.org/12/668812/3/gate/puppet-keystone-tripleo-standalone/8c7c376/ : SUCCESS in 1h 24m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a8c40e5d776c973fad798e3513fe22f114773d67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 17:21:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a60980f3da715b5d4ac986ae515e5aed88b35b74","revisions":{"4c93210b402bcf66e44ba8f7eaebcc39dc76ea53":{"kind":"REWORK","_number":1,"created":"2019-07-03 07:41:52.000000000","uploader":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"ref":"refs/changes/12/668812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/12/668812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/1"}}},"commit":{"parents":[{"commit":"5b5e713378f15d97c0c0c149f0d8807d87417bf7","subject":"Dont empty out a symlink","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5b5e713378f15d97c0c0c149f0d8807d87417bf7"}]}],"author":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-03 06:52:44.000000000","tz":180},"committer":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-03 06:52:44.000000000","tz":180},"subject":"Manage oslo_middleware/max_request_body_size","message":"Manage oslo_middleware/max_request_body_size\n\nSo that we can increase it from the default 114688\n\nUseful in case for example the OS-Federation mapping is too large.\n\nIf this limit is breached keystone will return a 413 Entity Too Large\nand not log anything to keystone.log.\n\nCloses-Bug: #1835161\n\nChange-Id: If9fc5c0bb5d6216b8656ee1673e1812c543de305\nSigned-off-by: Johan Guldmyr \u003cjohan.guldmyr@csc.fi\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/4c93210b402bcf66e44ba8f7eaebcc39dc76ea53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/4c93210b402bcf66e44ba8f7eaebcc39dc76ea53"}]},"branch":"refs/heads/master"},"0c9ddff12276d0e9c009f83411a5d1acb7811874":{"kind":"REWORK","_number":2,"created":"2019-07-03 08:05:34.000000000","uploader":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"ref":"refs/changes/12/668812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/12/668812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/2"}}},"commit":{"parents":[{"commit":"5b5e713378f15d97c0c0c149f0d8807d87417bf7","subject":"Dont empty out a symlink","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5b5e713378f15d97c0c0c149f0d8807d87417bf7"}]}],"author":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-03 06:52:44.000000000","tz":180},"committer":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-03 08:05:10.000000000","tz":180},"subject":"Manage oslo_middleware/max_request_body_size","message":"Manage oslo_middleware/max_request_body_size\n\nSo that we can increase it from the default 114688\n\nUseful in case for example the OS-Federation mapping is too large.\n\nIf this limit is breached keystone will return a 413 Entity Too Large\nand not log anything to keystone.log.\n\nCloses-Bug: #1835161\n\nChange-Id: If9fc5c0bb5d6216b8656ee1673e1812c543de305\nSigned-off-by: Johan Guldmyr \u003cjohan.guldmyr@csc.fi\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0c9ddff12276d0e9c009f83411a5d1acb7811874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0c9ddff12276d0e9c009f83411a5d1acb7811874"}]},"branch":"refs/heads/master"},"a60980f3da715b5d4ac986ae515e5aed88b35b74":{"kind":"REWORK","_number":3,"created":"2019-07-03 08:59:01.000000000","uploader":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"ref":"refs/changes/12/668812/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/12/668812/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/12/668812/3"}}},"commit":{"parents":[{"commit":"5b5e713378f15d97c0c0c149f0d8807d87417bf7","subject":"Dont empty out a symlink","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5b5e713378f15d97c0c0c149f0d8807d87417bf7"}]}],"author":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-03 06:52:44.000000000","tz":180},"committer":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-03 08:57:46.000000000","tz":180},"subject":"Manage oslo_middleware/max_request_body_size","message":"Manage oslo_middleware/max_request_body_size\n\nSo that we can increase it from the default 114688\n\nUseful in case for example the OS-Federation mapping is too large.\n\nIf this limit is breached keystone will return a 413 Entity Too Large\nand not log anything to keystone.log.\n\nCloses-Bug: #1835161\n\nChange-Id: If9fc5c0bb5d6216b8656ee1673e1812c543de305\nSigned-off-by: Johan Guldmyr \u003cjohan.guldmyr@csc.fi\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/a60980f3da715b5d4ac986ae515e5aed88b35b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/a60980f3da715b5d4ac986ae515e5aed88b35b74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
