)]}'
{"id":"openstack%2Fpuppet-keystone~682110","triplet_id":"openstack%2Fpuppet-keystone~master~Id0bc07f352d0b545e60aabd4523536dfc7fc59a8","project":"openstack/puppet-keystone","branch":"master","topic":"osc_train","hashtags":[],"change_id":"Id0bc07f352d0b545e60aabd4523536dfc7fc59a8","subject":"Replace url with endpoint in credentials provider","status":"MERGED","created":"2019-09-13 16:43:49.000000000","updated":"2019-09-20 04:34:34.000000000","submitted":"2019-09-20 04:34:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682110-1568954074004-ab427080","meta_rev_id":"78c099fbea987bb2acbeff2f91e17dce9364fa34","_number":682110,"virtual_id_number":682110,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-09-20 04:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-20 01:42:44.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-19 13:15:53.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-09-19 16:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-09-20 00:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-09-19 18:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-09-19 19:43:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 13:15:53.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-19 16:34:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-09-19 19:43:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-20 00:44:16.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-09-20 01:42:44.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-20 04:34:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1ee41002783304012764c691564382667f2b667","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-13 16:43:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b352106523569d33289e31fd9ea84ed8287543e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-13 17:02:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/10/682110/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/60a95ee/ : FAILURE in 16m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/10/682110/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/6eef6fb/ : FAILURE in 14m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/10/682110/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/2e76bc2/ : FAILURE in 14m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/10/682110/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/70cc35c/ : FAILURE in 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3eeef680883c3f8a4fbb875df8a523fbd9b1f2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 18:01:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05d2bde058e445bb845250f05068b07f : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2e059b5e9b84e79bfbf0107e8a738b3 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0434ce789b7f4d6a8fcf768c4c2b5fb6 : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a2d0cd3d2eb4454c931497008b0e91de : SUCCESS in 15m 54s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/671629de73e74793b14fd7e9de4a9f71 : SUCCESS in 11m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9a2697876a34104abe1d366932ef982 : FAILURE in 11m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/ef10152223b14cd883d821962dcf33d9 : FAILURE in 23m 53s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3d6fbebf219949a685624fa83baf3ba3 : FAILURE in 25m 43s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/34b0d34125724f5d83622868d7d06e14 : FAILURE in 22m 13s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/b598f5bae21e4b61b0a206751ef18c38 : FAILURE in 20m 43s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d4d68d5340bd46b09065cf60a1c19217 : FAILURE in 17m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c65fb71ce6194849bcbd17106694a848 : FAILURE in 15m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3a132f7d7174181a04b33c8ddd08dd5 : FAILURE in 21m 45s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9d33af39b18b4efaa549e170af283518 : FAILURE in 20m 32s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/22a8d0035e5b4806a749667be47d36c6 : FAILURE in 25m 10s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1814b5a7ad2a45cf8831a1c11bda9f5b : FAILURE in 23m 37s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9e78f2669e164b1e8395f981bb7b08ae : FAILURE in 1h 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"219a30d9a50f7cb742926264df72124dfd8d0d5f","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-16 08:44:16.000000000","message":"Patch Set 1: Workflow-1\n\nI think the integration test failures are happening because we still have older versions of openstackclient, and it looks like using --os-endpoint/--os-token with those versions is not working. I\u0027ll need to check.","accounts_in_message":[],"_revision_number":1},{"id":"e84f54cccf9192545f25db9b5d2544c6b6803268","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-16 13:16:41.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0411191fe950011e2cedc33cbd50b308702bc41b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 13:34:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/10/682110/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/512ca5d/ : FAILURE in 14m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/10/682110/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/b0ebaa8/ : FAILURE in 15m 00s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/10/682110/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/340f3bb/ : FAILURE in 15m 14s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/10/682110/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/06948d5/ : FAILURE in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b85f9b940576b644a57b3ba08f9680c8fb69ae8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 14:49:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a575b43b2af648bab975e1bea423152f : SUCCESS in 3m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e33f790c8b04ea6b3b9d40d4e8d7714 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef92c86e21bc4089b8ce961b6435a7e8 : SUCCESS in 4m 18s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ac8ff7db90194a2f9e802fdb9a800747 : SUCCESS in 14m 43s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e390d37253b44e62a2268ebad3601d6c : SUCCESS in 13m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8754d57cd4ee41d6bc1c16a4ffafad6a : FAILURE in 10m 45s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c4535eca91fc4fe4863baf6d2509ceb8 : FAILURE in 32m 11s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b6216c0d6f4b4711a32d1a632b5cbdf2 : FAILURE in 28m 51s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/82097f1da72b4db5a5d0a3090ec29ba3 : FAILURE in 26m 36s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/d2ce14fad806499fb8cdd5801e2d75dd : FAILURE in 23m 25s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/020db2f9dbc441c9b5a21a8e4d188ea4 : FAILURE in 26m 35s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ef16697eb99420b822f02786124be88 : FAILURE in 22m 43s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6b3430da1d14160929a8e90ddc8ef6f : FAILURE in 20m 26s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b93a022949d745bf9b7b10008364456e : FAILURE in 15m 26s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d8f1c1d4d35341a49ee053bc787a8ef1 : FAILURE in 23m 45s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a104b388421341ccac286dc6a80fd757 : FAILURE in 19m 36s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d350084d83b74e8793877ce29e759626 : FAILURE in 1h 26m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81c981de1f3710bd6d70c323e0d6920a52abe1d7","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-16 16:51:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"88c7ce7e5182e51638c052a3c035b38b55ef7700","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 16:56:30.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"e2d88ebd091d0386ccf420b4e7c532a51124c717","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 16:56:32.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"43698fe37703b580c7ed02417b27a9166c777662","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-16 16:56:51.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e30ce988129bf9ae7c0b4456acea2ac10922bf14","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 17:56:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/10/682110/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/79f816c/ : SUCCESS in 53m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/10/682110/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/7508a9e/ : SUCCESS in 57m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/10/682110/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/9d55b7f/ : SUCCESS in 58m 14s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/10/682110/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/3715507/ : SUCCESS in 45m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c2ce7526de716432dfaa6df42dfbc34258f18bff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:11:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5def9c7702ed4f0a9b6c25c623a76456 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c6d4ebd08b746e3b26b656186aab01a : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cb087fafc754f8981ffe5dc697f2886 : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5d47fc65f19c460ab77781d06c72b2e4 : SUCCESS in 24m 49s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cafc7dc9a9e54de48be2b17b28554e03 : SUCCESS in 15m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3496b21171c4b568a5dfeeb85d69ed9 : FAILURE in 13m 00s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a6a3ff26954a4c58ac9101c5257b3602 : SUCCESS in 1h 26m 31s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e4e81d0c29354ed6839eddc7d2bf5d2c : SUCCESS in 1h 31m 14s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/24d0ace2b0104f1f91a680f84209dad2 : SUCCESS in 1h 31m 54s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6fea074ffcd945b89fcc59fca8020dd8 : SUCCESS in 1h 04m 18s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fcdc9ede0c1440f6baf4a405397d4f4b : FAILURE in 53m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/664afcff61974ef4acf4fb6a50e5575a : FAILURE in 50m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8eef2f2499e47ab8fb26c9140bbf9bb : FAILURE in 40m 39s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/30ae038e4e054536ac5c28850dc08014 : FAILURE in 41m 32s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/916ab97f84734ab6ab7b72dbb43ed95a : SUCCESS in 1h 04m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/835bdb4379fc4c35bcb49b30003d6dba : FAILURE in 1h 09m 45s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ac06a93ce82544f184312a09e5b0ab68 : SUCCESS in 2h 07m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7facc1a74acd2ccc95433a4be2e62aedd8076ac8","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-09-17 11:16:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5f023422be493ff29213966543ed2e7a59d10758","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 12:07:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/10/682110/5/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/b1f1c23/ : SUCCESS in 44m 42s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/10/682110/5/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/7befd26/ : SUCCESS in 47m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/10/682110/5/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/694151f/ : SUCCESS in 49m 53s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/10/682110/5/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/5e8d00c/ : SUCCESS in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5857784d85a34574656191ab6053742104b6b2ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 13:04:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8463b21f8674898b69f3743b68095a0 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88cc76d9ff874aaaad202eec932569e0 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16e3c22535d44905a52d3d62bb7037a1 : SUCCESS in 4m 08s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a811df09565f42b2888303d192bf6493 : SUCCESS in 16m 56s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edea33929dd8484f8154ad1fe56bcad0 : SUCCESS in 10m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11dbc461a6e542a2a51e912d4cdbc9cb : FAILURE in 10m 54s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/ff371d965df14b47b879608ec9872c2c : SUCCESS in 51m 22s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3bda5ab4b80b4ffb8223f80e62fe7196 : SUCCESS in 57m 26s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3b99ea2f507142eab56a7b300541ed25 : SUCCESS in 1h 11m 08s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/fccd931b3d1e43ebbd2ca4f049380cd2 : SUCCESS in 45m 15s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4a598046948f4239ad745e85528d6501 : FAILURE in 37m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/329c91c6833f4eff90af3e0c24ea5484 : FAILURE in 31m 59s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f88e7cd1842481c88c68d6c6513f00e : FAILURE in 22m 27s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/33ccefdeb30047f3847380c80648ca58 : FAILURE in 22m 34s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f54e2f36d0b744ca9e9fd9b2ca81f6a4 : SUCCESS in 42m 23s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ede4f32c693f447c94740eb0b0b8b428 : FAILURE in 45m 30s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/068d338fa9934314862263603e7c44be : SUCCESS in 1h 40m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3c067b0836585b88aec16bdb69cbaa86dce7d5a8","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-19 13:15:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"cf16a6b03059ffcc569478b2dc149b6499693311","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-19 16:34:19.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"cbeae3bc914744125f99822d702aadc6ac703235","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-19 17:12:41.000000000","message":"Topic set to osc_train","accounts_in_message":[],"_revision_number":5},{"id":"ded0eeffe616f0686619fec5b3aee983c44507c2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-19 18:20:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"453c2f378ac30ebf93c1c139d91a086e72d1bd36","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-19 19:43:11.000000000","message":"Removed Workflow+1 by Alex Schultz \u003caschultz@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":5},{"id":"2dcabd74af746915141dcd04a3984d9356995d6b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-19 19:43:12.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bca9fa07e0b26f5e8b547d773b4975c0b0443c90","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-20 00:44:16.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f3777e8046cc7cdf2d37928ef5460bd31291e9d1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 01:42:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/10/682110/5/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/1651773/ : SUCCESS in 34m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/10/682110/5/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/b8ceabd/ : SUCCESS in 38m 49s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/10/682110/5/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/f771760/ : SUCCESS in 56m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/10/682110/5/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/8b5a2af/ : SUCCESS in 46m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d3ee9a24313a931b5166c97a77f3dc97fa614596","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 02:35:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b17b6fc3a7b4d2587f836416d8760eb : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a2af209e71b4baeb5b1fb730265824e : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d877ba764d6f47fd873cfb490ecf42b7 : SUCCESS in 5m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f1ffb7433faf43a5b5544ae2d94fdb34 : SUCCESS in 22m 00s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00480f19bd4542d2a17cafe7cdbd9d61 : SUCCESS in 13m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0ef5ca4e4df4d3e91ae4babaeec0c87 : FAILURE in 14m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/580411281a9b45718ba49b4cbfe96fd7 : SUCCESS in 1h 00m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/15d1e0f0a45541f3b7d79c93eac60abf : SUCCESS in 58m 25s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/85ea659fd88141569f84eb191c384ad4 : SUCCESS in 1h 07m 56s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/1bcc8605a9944c768b345a22b1f73256 : SUCCESS in 48m 47s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b10fe3f69aa147f186cbea8070c6b6e0 : FAILURE in 32m 36s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/858f541517fc4870b54c8933f33fb6eb : FAILURE in 29m 38s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83d22722b5ed4829b0d06abac70920d4 : FAILURE in 25m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ad0630509b964ed497bc260f0e21b1fe : FAILURE in 25m 01s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d20328abf6874a56b4369efd9101d7fc : SUCCESS in 37m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1194f4577f454304b2304d8d4ac346d4 : FAILURE in 51m 12s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f924a18cf97e40ef947deac12a7716bf : SUCCESS in 1h 47m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f7544463024041b89e677d29c432cfbfc86c92b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 02:35:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"52194d9a2f4d1153441a4fbf2beb6af8ca13550c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 04:34:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47163aba06bd43f290d7efdc3b1710b5 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb3a692009064ac88d413206e5cdd7fc : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4da6396cd182409091a620326f59aea3 : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/347c4eded5364dc59df3552ee722fb4b : SUCCESS in 20m 24s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9f338c6039d4f348a04e7bc09e1472c : SUCCESS in 14m 33s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c95045484ea845808a85b7eb92e194f9 : SUCCESS in 54m 59s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8c463868bac746dd92b63c9fb19a0644 : SUCCESS in 1h 03m 08s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c77db379e33641149fdbd2b0512432c3 : SUCCESS in 1h 10m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/1d2b9e0ec33f4439b8418263699a5f0d : SUCCESS in 47m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/49122397bcf549e6b20b7ab1facf91dc : SUCCESS in 37m 51s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7b8f8543b8704006b4e2d965ccdb55c1 : SUCCESS in 1h 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"78c099fbea987bb2acbeff2f91e17dce9364fa34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 04:34:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9922e4196f6fc2ec584b8d3f3ea652044bb32da9","revisions":{"86b3e246ad2bffdc180fd4efbc11055c556d0e3d":{"kind":"REWORK","_number":1,"created":"2019-09-13 16:43:49.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/10/682110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/10/682110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/1"}}},"commit":{"parents":[{"commit":"7d4c760446f81916895867d1825a44d8fd30a8e5","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7d4c760446f81916895867d1825a44d8fd30a8e5"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-09-13 16:38:29.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-09-13 16:43:47.000000000","tz":120},"subject":"[WIP] Replace url with endpoint in credentials provider","message":"[WIP] Replace url with endpoint in credentials provider\n\npython-openstackclient has removed the --os-url option in [1]. This\ncommit is part of the upcoming 4.0.0 release.\n\nThe openstack provider relies on that for Keystone initial configuration,\nso we need to fix this or puppet-keystone will be broken.\n\n[2] is a more solid long-term solution, but this can work as a stop-gap.\n\n[1] - https://review.opendev.org/677795\n[2] - https://review.opendev.org/630714\n\nDepends-On: https://review.opendev.org/682108\nChange-Id: Id0bc07f352d0b545e60aabd4523536dfc7fc59a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/86b3e246ad2bffdc180fd4efbc11055c556d0e3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/86b3e246ad2bffdc180fd4efbc11055c556d0e3d"}]},"branch":"refs/heads/master"},"17af507a20860fa9d2ac19475983f70fa737a9ed":{"kind":"REWORK","_number":2,"created":"2019-09-16 13:16:41.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/10/682110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/10/682110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/2"}}},"commit":{"parents":[{"commit":"7d4c760446f81916895867d1825a44d8fd30a8e5","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7d4c760446f81916895867d1825a44d8fd30a8e5"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-09-13 16:38:29.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-09-16 13:16:37.000000000","tz":0},"subject":"[WIP] Replace url with endpoint in credentials provider","message":"[WIP] Replace url with endpoint in credentials provider\n\npython-openstackclient has removed the --os-url option in [1]. This\ncommit is part of the upcoming 4.0.0 release.\n\nThe openstack provider relies on that for Keystone initial configuration,\nso we need to fix this or puppet-keystone will be broken.\n\n[2] is a more solid long-term solution, but this can work as a stop-gap.\n\n[1] - https://review.opendev.org/677795\n[2] - https://review.opendev.org/630714\n\nDepends-On: https://review.opendev.org/682108\nChange-Id: Id0bc07f352d0b545e60aabd4523536dfc7fc59a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/17af507a20860fa9d2ac19475983f70fa737a9ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/17af507a20860fa9d2ac19475983f70fa737a9ed"}]},"branch":"refs/heads/master"},"d935d7f961e1d2b2283c4e13709f8e6dfb94122e":{"kind":"REWORK","_number":3,"created":"2019-09-16 16:51:00.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/10/682110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/10/682110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/3"}}},"commit":{"parents":[{"commit":"7d4c760446f81916895867d1825a44d8fd30a8e5","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7d4c760446f81916895867d1825a44d8fd30a8e5"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-09-13 16:38:29.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-09-16 16:50:57.000000000","tz":0},"subject":"[WIP] Replace url with endpoint in credentials provider","message":"[WIP] Replace url with endpoint in credentials provider\n\npython-openstackclient has removed the --os-url option in [1]. This\ncommit is part of the upcoming 4.0.0 release.\n\nThe openstack provider relies on that for Keystone initial configuration,\nso we need to fix this or puppet-keystone will be broken.\n\n[2] is a more solid long-term solution, but this can work as a stop-gap.\n\n[1] - https://review.opendev.org/677795\n[2] - https://review.opendev.org/630714\n\nDepends-On: https://review.opendev.org/682108\nChange-Id: Id0bc07f352d0b545e60aabd4523536dfc7fc59a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d935d7f961e1d2b2283c4e13709f8e6dfb94122e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d935d7f961e1d2b2283c4e13709f8e6dfb94122e"}]},"branch":"refs/heads/master"},"a5d616cdae47f3850ecdde73b90105cce76d0d14":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-09-16 16:56:51.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/10/682110/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/10/682110/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/4"}}},"commit":{"parents":[{"commit":"7d4c760446f81916895867d1825a44d8fd30a8e5","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7d4c760446f81916895867d1825a44d8fd30a8e5"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-09-13 16:38:29.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-09-16 16:56:49.000000000","tz":0},"subject":"[WIP] Replace url with endpoint in credentials provider","message":"[WIP] Replace url with endpoint in credentials provider\n\npython-openstackclient has removed the --os-url option in [1]. This\ncommit is part of the upcoming 4.0.0 release.\n\nThe openstack provider relies on that for Keystone initial configuration,\nso we need to fix this or puppet-keystone will be broken.\n\n[2] is a more solid long-term solution, but this can work as a stop-gap.\n\n[1] - https://review.opendev.org/677795\n[2] - https://review.opendev.org/630714\n\nDepends-On: https://review.opendev.org/682108\nDepends-On: https://review.opendev.org/682415\nChange-Id: Id0bc07f352d0b545e60aabd4523536dfc7fc59a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/a5d616cdae47f3850ecdde73b90105cce76d0d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/a5d616cdae47f3850ecdde73b90105cce76d0d14"}]},"branch":"refs/heads/master"},"9922e4196f6fc2ec584b8d3f3ea652044bb32da9":{"kind":"REWORK","_number":5,"created":"2019-09-17 11:16:02.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/10/682110/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/10/682110/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/10/682110/5"}}},"commit":{"parents":[{"commit":"7d4c760446f81916895867d1825a44d8fd30a8e5","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7d4c760446f81916895867d1825a44d8fd30a8e5"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-09-13 16:38:29.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-09-17 11:15:58.000000000","tz":120},"subject":"Replace url with endpoint in credentials provider","message":"Replace url with endpoint in credentials provider\n\npython-openstackclient has removed the --os-url option in [1]. This\ncommit is part of the upcoming 4.0.0 release.\n\nThe openstack provider relies on that for Keystone initial configuration,\nso we need to fix this or puppet-keystone will be broken.\n\n[2] is a more solid long-term solution, but this can work as a stop-gap.\n\n[1] - https://review.opendev.org/677795\n[2] - https://review.opendev.org/630714\n\nDepends-On: https://review.opendev.org/682108\nDepends-On: https://review.opendev.org/682415\nChange-Id: Id0bc07f352d0b545e60aabd4523536dfc7fc59a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/9922e4196f6fc2ec584b8d3f3ea652044bb32da9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/9922e4196f6fc2ec584b8d3f3ea652044bb32da9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
