)]}'
{"id":"openstack%2Fpuppet-keystone~712858","triplet_id":"openstack%2Fpuppet-keystone~master~Idf259de1c9599568ed39774f3b0e6e060310873f","project":"openstack/puppet-keystone","branch":"master","topic":"centos-8-jobs","hashtags":[],"change_id":"Idf259de1c9599568ed39774f3b0e6e060310873f","subject":"Use centos-8 standalone puppet parent","status":"MERGED","created":"2020-03-13 08:47:51.000000000","updated":"2020-03-16 15:40:49.000000000","submitted":"2020-03-16 15:40:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712858-1584373249504-74ea99eb","meta_rev_id":"34f8ca7ede892fcbe6514e1f7e4321d3440c8464","_number":712858,"virtual_id_number":712858,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-16 15:40:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-13 22:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-03-16 14:09:15.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 22:21:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-16 14:09:15.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-16 15:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b3fa4f9ccd2a5a595e9464080dfa7abdd1a9697","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-03-13 08:47:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86bac504a98f062f98c633f052feeb5da29449a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 10:33:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd3c7174a4064f6d8fe1502b51204510 : SUCCESS in 5m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a30b3f691b9e47e596aea131eb7a2a66 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd9e019cbcce49818319f2b4995b729f : SUCCESS in 7m 20s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8b8fa51f87554f4c819f4ccefae685f0 : SUCCESS in 23m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/760be31a09e84551875a2a2a5dfc8533 : SUCCESS in 13m 58s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2e96a00dc14743b28b760b12260d9b3a : SUCCESS in 19m 27s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96e795d97b9047f3b8a7f9a8365e1d32 : SUCCESS in 15m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6dc1a3f2a9ed455ea637c3b87fa75e43 : FAILURE in 12m 53s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/27142dde335e4ad684fc0b9f7e6043f6 : FAILURE in 1h 09m 50s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d2c28e1ad252483c9480c468009aa009 : SUCCESS in 1h 16m 40s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/77f498c301aa4eaabd361119627e3d87 : SUCCESS in 1h 05m 16s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/f05458d9eb2a4b859f878a42c919d066 : SUCCESS in 46m 46s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2a690411f70245b2a103c253ec339de9 : SUCCESS in 45m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfa4177c5a0f4852bcd87f046e3cc934 : SUCCESS in 58m 00s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/688733d0f6ac41b8b428b7632cc54f4b : SUCCESS in 36m 01s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/48be0f377cf7438cb800204d61ed8632 : SUCCESS in 39m 38s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2541eebac9f94b70b83c127d5836db25 : SUCCESS in 59m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cc9055a2e7b74a95bcbf6fbeab41836c : SUCCESS in 1h 18m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4407772852724a52b2d68ae86aade646 : SUCCESS in 57m 33s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/de306d483ad74105b7d366eece16cc67 : SUCCESS in 1h 01m 23s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7a43eee7fc9e47288299a898a4e66bef : SUCCESS in 43m 07s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dea9e32f973a4428a8cacbdf5be3c112 : FAILURE in 58m 31s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2cb0cfc4ba7b401ba9af5779e48b29fc : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"216eb544fded6166427c19e4b35e0667bb7e176c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-13 20:14:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5eec86298cd4ca94d617ac52f413cc7ddfcf27c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:03:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/400a74f6b220460097bd4bf6722e4ef0 : SUCCESS in 5m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a82a3130d5140279166d8ae292257f8 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a83235b2becc4dc5be33972b5d4d2733 : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/71f8216577af4b5f92cfc0766cc33f52 : SUCCESS in 16m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5ea688808aa745eda1eebe8a018e69f9 : SUCCESS in 14m 18s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/7b34b399b74441ee8c3fbad540adcf3e : SUCCESS in 14m 54s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/947d6cef8572443f80a1a2572d712877 : SUCCESS in 12m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/773c59e1eed3418cb448e802ae6bee15 : FAILURE in 13m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/37e6bfd9fb754159946a46afd99db47e : SUCCESS in 57m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/67ae68493e674e3ca46942a9ac23ac5a : SUCCESS in 1h 05m 58s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1219d335610a476d972598f0d40ab8c9 : SUCCESS in 1h 06m 43s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/882c7ed3ecf94712883525636cc91efe : SUCCESS in 51m 23s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/936e5a92d97848d28e76c26a264c782d : SUCCESS in 45m 52s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f09122bc238a4a8daaf4fafe51541bb6 : SUCCESS in 50m 57s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78d79d8c89864aa398bf176e40a081f9 : SUCCESS in 39m 45s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d51598f869f04e238aecee214aeab8fc : SUCCESS in 40m 45s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f4842b9c5d8445578d195e0cbe43850d : SUCCESS in 1h 05m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/86bd122b7c0448aa88514f05ebcb4638 : SUCCESS in 1h 08m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a878641526f9451ca851fc5ab9f6796c : SUCCESS in 1h 19m 10s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1de5e7f01b2742539e1a705ea4577411 : SUCCESS in 43m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9f3dadfa8c204c34ae89fd4884db6122 : SUCCESS in 48m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/679081bf15be46b0adced8f977fb8a04 : FAILURE in 59m 26s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ddc663476394da392dd553b14ccbd0c : SUCCESS in 1h 43m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8d2e53e0e69385559551f4226d96d859684a989f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-13 22:21:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"76a43c6c073c2cfa845eaeb1f3e5e638071a0e66","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-16 14:09:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a44c11ac58d9466cf9c116eb0f1451f459ab1bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 14:09:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a92cd03e17d17543227b6b0bd91f15ac5be1910f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 15:40:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6a12cb91031bf4dd40c386031726dc930b28d24b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 15:40:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/979b8491260a48108653c38a103259a5 : SUCCESS in 8m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed65027cf439431496a0ac9be2182e9b : SUCCESS in 7m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12ed02a6a34c48e4bb4f2a025dbc74d4 : SUCCESS in 6m 22s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b260783ac51949a18dec5bb7d592e32e : SUCCESS in 17m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8bb2ac24330479f980f92bfa7667fbf : SUCCESS in 16m 27s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a712fab4add24ada9d0c40fc1b315fea : SUCCESS in 1h 05m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/059435b811ae44ba875478cd394f72cf : SUCCESS in 1h 19m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/05ed9405fc374383bcc90c133f204c8c : SUCCESS in 58m 45s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/cb9c6c461f1e45d59999219e0fa4f877 : SUCCESS in 49m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9d933a90b14d4084bd963f5b47290c3d : SUCCESS in 40m 42s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/162dce24af9843a5aa12cbde210d0ab8 : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad4c243e11f9fc678bf3cb238cba754febc8e29d","revisions":{"ad4c243e11f9fc678bf3cb238cba754febc8e29d":{"kind":"REWORK","_number":1,"created":"2020-03-13 08:47:51.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/58/712858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/58/712858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/58/712858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/58/712858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/58/712858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/58/712858/1"}}},"commit":{"parents":[{"commit":"7586bc88c15ffb3383324befffee37fda103ee2e","subject":"Merge \"Fix performance regression due to reduced number of keystone workers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7586bc88c15ffb3383324befffee37fda103ee2e"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-03-13 08:44:18.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-03-13 08:46:33.000000000","tz":330},"subject":"Use centos-8 standalone puppet parent","message":"Use centos-8 standalone puppet parent\n\nhttps://review.opendev.org/#/c/711320/ adds\ntripleo-puppet-ci-centos-8-standalone job in poi and let\u0027s\nre-use the same parent for tripleo standalone job.\n\nChange-Id: Idf259de1c9599568ed39774f3b0e6e060310873f\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/ad4c243e11f9fc678bf3cb238cba754febc8e29d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/ad4c243e11f9fc678bf3cb238cba754febc8e29d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
