)]}'
{"id":"openstack%2Fpuppet-keystone~721127","triplet_id":"openstack%2Fpuppet-keystone~stable%2Fqueens~I507d1b736dbbb147c67b9d399c033703b432b16d","project":"openstack/puppet-keystone","branch":"stable/queens","hashtags":[],"change_id":"I507d1b736dbbb147c67b9d399c033703b432b16d","subject":"Update ldap-backend options","status":"MERGED","created":"2020-04-19 23:57:43.000000000","updated":"2020-04-20 20:43:21.000000000","submitted":"2020-04-20 20:43:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721127-1587415401161-dfe5b5ea","meta_rev_id":"4b45dfcecf61abb92cdfbfd9a74d639bfdf4ed43","_number":721127,"virtual_id_number":721127,"owner":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 20:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-20 19:30:18.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2020-04-20 14:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-20 19:30:18.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-04-19 23:57:43.000000000","_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-19 23:57:43.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-04-20 14:44:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-20 19:30:18.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-20 20:43:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03ef39b4dc73359b2d8b3432f6e6ee66fdde45e5","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2020-04-19 23:57:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"23a96a1b61ec40195024c7a17388f23b3726dcc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 01:19:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/39fe213fcd634e9cb1acc152fb3a4fd0 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/90b580d92eeb44d5a4a4a3c81a5763db : SUCCESS in 4m 10s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/ddfe92a8ba264150a1e889693211c01c : SUCCESS in 17m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0f56b61ddc2f4b36abbb6a8f7499f173 : SUCCESS in 15m 00s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b2952ab143894da9a87d9987fcf940e5 : SUCCESS in 59m 04s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ce64bcfc563042a785afa60c847954ae : SUCCESS in 1h 20m 29s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4f17b0275d5f46fca40ed705f8622a50 : SUCCESS in 58m 33s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/e0edec3b06284d0c921693dceb622159 : SUCCESS in 41m 01s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/806d8a30e94f407ebaa2a878a13db5f0 : FAILURE in 33m 05s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6ee3018f5fc848be96ad83be7d56791a : FAILURE in 27m 58s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b1db0cf6b6304c11b59a78c7078184d9 : FAILURE in 24m 11s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/c88c28d2769245c0ac512816b9f96cce : FAILURE in 22m 49s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0576bdb00edc4952b70c1457f36a4626 : SUCCESS in 36m 55s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/37c261889e804d1298ddcdaf26ad13ea : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"480a56ac8bef02bc57a0da891980fc41e38398d2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-20 14:44:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c313cf6942650ce5f75704a3b82556514198f7a0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-20 19:30:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d97774ecac958ae44818abab83ffd00164d1fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 19:30:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c10e567dc51d962ba731877f9165a3a95ff54aeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:43:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/4bb3b142d3704a148fe294afd87069ba : SUCCESS in 5m 13s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/53766b2c44dd49b5af51079e168a41b8 : SUCCESS in 6m 22s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/879db8f8045345edaba7612fd1109ca0 : SUCCESS in 21m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2652108ad1ac4d3a9625893dd0e1ec24 : SUCCESS in 13m 39s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/bb7ceffe7f77425d84faf685575c3a06 : SUCCESS in 1h 05m 21s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fa8d30c332d4425fb99134b5f94798f8 : SUCCESS in 1h 06m 31s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3a297bb9fd4c4761bbdf9c6e9ce7819d : SUCCESS in 1h 10m 33s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/e11a6f57b9a94fa4bd8631a1d94b9e3f : SUCCESS in 54m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a21f8f5c065449c5a9ceba9aabc95287 : SUCCESS in 35m 04s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/0d45f807d10c4acbbfe92f585cf52910 : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4b45dfcecf61abb92cdfbfd9a74d639bfdf4ed43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:43:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f65517dcf02145606c3c524c52d237639775a0e7","revisions":{"f65517dcf02145606c3c524c52d237639775a0e7":{"kind":"REWORK","_number":1,"created":"2020-04-19 23:57:43.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/27/721127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/27/721127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/27/721127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/27/721127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/27/721127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/27/721127/1"}}},"commit":{"parents":[{"commit":"86798f135ef7865dbdbc6a7d0f8413810af25184","subject":"Remove PE requirement from metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/86798f135ef7865dbdbc6a7d0f8413810af25184"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-03-18 18:04:58.000000000","tz":0},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2020-04-19 23:57:43.000000000","tz":0},"subject":"Update ldap-backend options","message":"Update ldap-backend options\n\nSeveral of the options defined in ldap_backend.pp were using duplicate\nvalues already specified in keystone/conf/ldap.py. Instead of\nduplicating the same value, we can set them to undefined and just let\nthe default values from keystone come through.\n\nThis commit also updates the values of use_pool and use_auth_pool to\nTrue so they\u0027re consistent with the default values in keystone.\n\nCo-Authored-By: Dave Wilde \u003cdwilde@redhat.com\u003e\n\nChange-Id: I507d1b736dbbb147c67b9d399c033703b432b16d\n(cherry picked from commit 1081ac51db7f560e5856809480fb98c215747476)\n(cherry picked from commit e59b90669832d5bc20da2da0c0ac5ed9c16e656f)\n(cherry picked from commit 2821509f0e7f67b711257d1621dfe28701b910aa)\n(cherry picked from commit 6e29624320cb4cb5ed3bcc4bc93cae609cd3effe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/f65517dcf02145606c3c524c52d237639775a0e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/f65517dcf02145606c3c524c52d237639775a0e7"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
