)]}'
{"id":"openstack%2Fpuppet-keystone~729213","triplet_id":"openstack%2Fpuppet-keystone~master~Ie564c3ae6023a36f317b963872059509754970e5","project":"openstack/puppet-keystone","branch":"master","topic":"db-password","hashtags":[],"change_id":"Ie564c3ae6023a36f317b963872059509754970e5","subject":"Remove password hash generation in each puppet modules","status":"MERGED","created":"2020-05-19 11:09:54.000000000","updated":"2020-05-20 05:40:17.000000000","submitted":"2020-05-20 05:40:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729213-1589953217156-f8705a57","meta_rev_id":"4936e51d4b8ea5337de3fe3c59d55ce0b41cd608","_number":729213,"virtual_id_number":729213,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-20 02:34:22.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 05:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-19 12:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-05-19 21:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-05-19 21:47:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 12:12:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-19 21:47:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-20 05:40:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efed7c5e8967409b2f5e04b66b6b566e4f0e2bc6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 11:09:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c69b19e7454299101ccaa937e0e3b19f2d2ea935","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 11:14:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"341d1bbcbe1629783919a6726689dd36543b84ce","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-19 12:12:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d449c03803d7580070518e480300aedb66cc0fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 12:37:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/9efb49ef448642b7b9414fd5f17b0535 : SUCCESS in 3m 05s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0167f1c6ce42459e9e70128d67ae4bc2 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4750c7029a5441e86327b46dc582f1b : SUCCESS in 5m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86ddaa18224d434db48670a8c6edb47e : SUCCESS in 5m 35s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2dfc515b62a44c9d926b0610c89af97f : SUCCESS in 17m 46s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9d3de39046c34cd6a0a19742b83602e8 : SUCCESS in 15m 52s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08477588e8604cae88bb1455361af5a9 : SUCCESS in 14m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9e98ed9d8bb497cb297ed44e66bbf7c : FAILURE in 13m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5a7d0fe2a89a47f2a30b0d3d6af4a9d4 : FAILURE in 34m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18ab1bab93d94895a4fcb341cbb1243c : FAILURE in 33m 42s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab08fada28b549f98276cef4ec67e9b1 : FAILURE in 32m 40s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/61a419f13a9c45c681e18848135310c2 : FAILURE in 25m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bdbe288e74e148e79d22e2b683af3c70 : SUCCESS in 54m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3b70036d41f349b4b26c098b714db786 : SUCCESS in 1h 03m 10s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/965db57fa5f649419282d14e401211fd : SUCCESS in 1h 00m 29s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0610a8f0ffbd4dcaa22d98095398d594 : SUCCESS in 42m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/318c4f187ea94bfb949026af4685ffe0 : SUCCESS in 43m 54s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ec623a91b6ff4680b1f5071256465673 : FAILURE in 10m 12s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/060d891b4c6147409b783dfba737862d : FAILURE in 7m 14s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/23d6ff97b5744882abe0ea5d8bf5f9f5 : SUCCESS in 1h 21m 07s","accounts_in_message":[],"_revision_number":2},{"id":"22562be11b9eed8dd16ef30cc733f11e0786b14f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-19 21:47:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cc5e637ae514f46b2542d92ce54dffa26d95354f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-20 02:34:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2a14abd52ade23a3da5183e9a9beb82393a034be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 04:07:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/4cbcf2409b274fd897bce267ac34aa4a : SUCCESS in 4m 46s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aaffe7f7cc7340c58b021406c2c0d791 : SUCCESS in 6m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c937e0b853d45bd9eff8262764367cf : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87719ab9a33d4cef95ad819da3e04a22 : SUCCESS in 6m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8ada51d4f95646f6b659b47b6f9e118a : SUCCESS in 15m 07s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/71885cecf00a425ea499776cc77e9f78 : SUCCESS in 17m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23880d49c2c04947a2e7019b7b2ad964 : SUCCESS in 14m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d77f7c34064420a9c5038e95d867744 : FAILURE in 14m 20s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cff24050f75f4ac6acd62c7e8e29cd95 : FAILURE in 35m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97cd0e0595684fcd9d4780a6aa812cd3 : FAILURE in 42m 01s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/313809723c1a4d819124f57de698fc48 : FAILURE in 36m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1a30121424524a089ad6bdf7bf035dc1 : FAILURE in 26m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d1168140e37647ccbcc10c1c0ab4dfcd : SUCCESS in 56m 51s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1c819159787540c1b4ba21966252903b : SUCCESS in 1h 04m 45s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/86eb19124be34026aa962fa7c75a08f0 : SUCCESS in 1h 11m 48s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/06dd150e53794125bcae2f947a6e26dc : SUCCESS in 46m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/10cb6de3caa2454aac9abb802aea09ef : SUCCESS in 46m 16s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4a8fe641bd6d465c940f755b2866fd35 : FAILURE in 9m 34s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38e9c57ceddf4f83b3afa91b618e869d : FAILURE in 11m 24s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4e113b6d921a4c1b99b327d60014dd35 : SUCCESS in 1h 28m 24s","accounts_in_message":[],"_revision_number":2},{"id":"1c4fd8daa14f47dc0d67c3d60718e6522efb26dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 04:07:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"043004e886f3df1e5a039ae8615b69fbcb2913ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 05:40:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/7a5c84b09a1b4086b47a7e2addcd0ead : SUCCESS in 2m 24s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aaf8df732d3140439a589b62fe03229f : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee7b9bc1efa04427914752d94f065772 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9485a3bc1584ee7ac615e83cbb21206 : SUCCESS in 4m 33s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8d85532b37e142798b32869f93d8e015 : SUCCESS in 14m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/7071ac7f21d24e3396f01749f861f05d : SUCCESS in 16m 06s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c08108489d340e9bc8b5d965b1d5eef : SUCCESS in 12m 33s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e0d850e1483043c6815cc0e9cc195ebe : SUCCESS in 55m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e94ab3f459ab46d083aa638bb2355244 : SUCCESS in 1h 00m 23s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2c385114282546e0a8cffa5f6bdc3d01 : SUCCESS in 1h 09m 58s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e311be8e3699468ab527231e51fc5b83 : SUCCESS in 47m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/17a7d8c8b6d646c68fe0d13c0db821d1 : SUCCESS in 40m 35s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b33518c1699f400a88690c56b1da7ba2 : SUCCESS in 1h 28m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5b662c9fec24b43d74648f70e97243817719f512","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 05:40:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ab7ef40448f93be1f9a9dfd2fc2bd068481bb97","revisions":{"3bbf2640bfef78b9eb4b291dbfc2962f10748fab":{"kind":"REWORK","_number":1,"created":"2020-05-19 11:09:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/729213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/13/729213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/13/729213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/13/729213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/13/729213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/13/729213/1"}}},"commit":{"parents":[{"commit":"368683a8d0606988e8cf2a4bb9657a2782a9135e","subject":"Merge \"[ussuri][goal] Add CONTRIBUTING.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/368683a8d0606988e8cf2a4bb9657a2782a9135e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 11:09:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 11:09:38.000000000","tz":540},"subject":"Remove password hash generation in each puppet modules","message":"Remove password hash generation in each puppet modules\n\n... and migrate it to openstacklib so that all logics about database\nconfiguration are implemented in one common place.\n\nDepends-on: https://review.opendev.org/#/c/728595/\nChange-Id: Ie564c3ae6023a36f317b963872059509754970e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3bbf2640bfef78b9eb4b291dbfc2962f10748fab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3bbf2640bfef78b9eb4b291dbfc2962f10748fab"}]},"branch":"refs/heads/master"},"8ab7ef40448f93be1f9a9dfd2fc2bd068481bb97":{"kind":"REWORK","_number":2,"created":"2020-05-19 11:14:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/729213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/13/729213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/13/729213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/13/729213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/13/729213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/13/729213/2"}}},"commit":{"parents":[{"commit":"368683a8d0606988e8cf2a4bb9657a2782a9135e","subject":"Merge \"[ussuri][goal] Add CONTRIBUTING.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/368683a8d0606988e8cf2a4bb9657a2782a9135e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 11:09:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 11:14:26.000000000","tz":540},"subject":"Remove password hash generation in each puppet modules","message":"Remove password hash generation in each puppet modules\n\n... and migrate it to openstacklib so that all logics about database\nconfiguration are implemented in one common place.\n\nDepends-on: https://review.opendev.org/#/c/728595/\nChange-Id: Ie564c3ae6023a36f317b963872059509754970e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8ab7ef40448f93be1f9a9dfd2fc2bd068481bb97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8ab7ef40448f93be1f9a9dfd2fc2bd068481bb97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
