)]}'
{"id":"openstack%2Fpuppet-keystone~731134","triplet_id":"openstack%2Fpuppet-keystone~master~I51184bb0153f2754dc6bc0d445020d218f6ecc3a","project":"openstack/puppet-keystone","branch":"master","hashtags":[],"change_id":"I51184bb0153f2754dc6bc0d445020d218f6ecc3a","subject":"Install the correct memcache bindings for py3","status":"MERGED","created":"2020-05-27 10:06:40.000000000","updated":"2020-05-27 21:47:07.000000000","submitted":"2020-05-27 21:45:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731134-1590615946471-41f9ad85","meta_rev_id":"70b5dad047ca54f98cf078ad6d59bd51ba04409a","_number":731134,"virtual_id_number":731134,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 21:45:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-27 14:59:49.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-27 20:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-05-27 19:21:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-05-27 20:07:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 14:59:49.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-27 19:21:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-27 20:07:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-27 21:45:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed1247d907527d0deeb469ce08e64575415fe688","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-27 10:06:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56870b8ca3874a3fe600d123d6af3bf6c26aa767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:43:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1947ff9617914a57ade26e223a93d750 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c75ceb292c164f30b8bd3d8af86aa1e9 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/332a3055b07b4aceaf12f00bd0f55b06 : SUCCESS in 6m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1415d5bc77c14188a77a25ff9ccd020a : SUCCESS in 17m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3b717109c1a1485ba6c1ece71962a85b : SUCCESS in 18m 42s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d37ba08096d04fdcb2e82dbfec5617cf : SUCCESS in 13m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4c709845ef64103b129237e982dc697 : FAILURE in 13m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/56b16c344c4a4927b9638baa81d82fb1 : FAILURE in 36m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/473b491e239f4a10a8526dd8e027c83a : FAILURE in 38m 10s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/732a9a51591c4631ba5de49707078092 : FAILURE in 33m 28s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b7cf9d026d43435bad65b33117eda661 : FAILURE in 41m 44s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1e7eac3354c84be2b8113428ec2f990f : SUCCESS in 1h 06m 11s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cd55e3b894f34a869307219bc9a9009a : SUCCESS in 1h 06m 06s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/44d71afe2d1b470994d003a8c3afe3aa : SUCCESS in 1h 14m 29s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/729cc1038d2444f58a3fb851d19d17f1 : SUCCESS in 55m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/dc55e8ac5bed4b48b5331cd5f0a1662e : SUCCESS in 47m 53s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/960f9a17512e433f9a9b356696dc5c7d : FAILURE in 12m 29s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3635cebd12d4c34ae827ca0a556d9b7 : FAILURE in 8m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bb864db7ba447239841163872633bf6 : SUCCESS in 3m 52s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/743c613397dd4daa95d45b06b2cc56e5 : SUCCESS in 1h 33m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8c4334c6c6648ef7d88426364b6a851980a954af","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-27 14:59:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02d36a476b67400c0bc452b241709c25a5119ca0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-27 20:07:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4be73bd6a933ae86efc291a5f6a45f36cd7c22a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 20:07:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b491205d9272874d222cef972afe4d038d798de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:45:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a953c81d6bca271c4afbe812e672f993538018ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:45:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4db9c1797a174a23b45f82fc46a990fb : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b93fd60056eb462b8469bd1c4e1f7d7b : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c31fbcef75884b7b937a64b084986696 : SUCCESS in 5m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/883291987c674ceebd9efd268c9d00f5 : SUCCESS in 18m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fdd55dd2d83547348ef0eab560691a98 : SUCCESS in 17m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdcf400550584e87a54c7c93cf6933a1 : SUCCESS in 14m 51s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5684c475880243149de5b5abcb2e0b31 : SUCCESS in 57m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/693bab83521a405fb80ef732fb84ace7 : SUCCESS in 1h 06m 05s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e994c1858a054d3dbf7e6ebe42236143 : SUCCESS in 1h 11m 15s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8bb63374710343aa819e63d9d81e7dd3 : SUCCESS in 45m 44s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b0b30236ffbb46779827d67dff5e9e31 : SUCCESS in 48m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36d7cc59c0be42ea9170406e94a946f7 : SUCCESS in 3m 45s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4e618188ed5640db977a18b7acdffe2a : SUCCESS in 1h 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"37f303538ae41d5bc16ddf1c78786bf0c33131d3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-27 21:47:02.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 975806ce18e655b3453556f59e64f3e46b70157b","accounts_in_message":[],"_revision_number":1},{"id":"70b5dad047ca54f98cf078ad6d59bd51ba04409a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:47:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79fbc1c51d6643d6b16756e0a1463ae5 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fa63db9387e7f0910d4e1138b053a5055066f89","revisions":{"3fa63db9387e7f0910d4e1138b053a5055066f89":{"kind":"REWORK","_number":1,"created":"2020-05-27 10:06:40.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/34/731134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/34/731134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/34/731134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/34/731134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/34/731134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/34/731134/1"}}},"commit":{"parents":[{"commit":"f4f18ec73bf7576d805f29985d54af66063087a1","subject":"Remove unnecessary puppet-rabbitmq from dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/f4f18ec73bf7576d805f29985d54af66063087a1"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-27 10:06:26.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-27 10:06:26.000000000","tz":120},"subject":"Install the correct memcache bindings for py3","message":"Install the correct memcache bindings for py3\n\nChange-Id: I51184bb0153f2754dc6bc0d445020d218f6ecc3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3fa63db9387e7f0910d4e1138b053a5055066f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3fa63db9387e7f0910d4e1138b053a5055066f89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
