)]}'
{"id":"openstack%2Fpuppet-keystone~739380","triplet_id":"openstack%2Fpuppet-keystone~master~I8a7abed59468f93d48aa38a47ad4f72d994da79a","project":"openstack/puppet-keystone","branch":"master","topic":"token-flush","hashtags":[],"change_id":"I8a7abed59468f93d48aa38a47ad4f72d994da79a","subject":"Deprecate keystone::cron::token_fush","status":"MERGED","created":"2020-07-05 23:42:05.000000000","updated":"2020-07-06 15:50:06.000000000","submitted":"2020-07-06 15:44:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739380-1594050283933-f3086ce0","meta_rev_id":"f4361511c0522121eb4ab8f2ccbf94074f744d90","_number":739380,"virtual_id_number":739380,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-07-06 03:30:54.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 15:44:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-07-06 14:03:09.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-07-06 08:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-07-06 14:03:09.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 08:57:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-07-06 14:03:09.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-06 15:44:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"879eba3e8f54a10e45141ade5db3db03e353448a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-05 23:42:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb5431db357671ced83612c76f250dd33a42b8b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 00:55:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc4add8962ee47ba957bcf526b944ff4 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d09889429634d72bbe7189c6cdc3a90 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ab3a32d482a4e02bf9491e893864284 : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/484da188e92e43adaef5288b6f5d0940 : SUCCESS in 12m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/45dd1ba78d734a72bc01c546d9967ce0 : SUCCESS in 15m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7907015b15c4ad5ae89980696fe40ca : SUCCESS in 13m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8775756493a647119b560fb1c605be88 : FAILURE in 11m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c2898fe882bf4341a12f71c96b403e67 : FAILURE in 5m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9afc1f0585e14621af24ca04cacb4d9a : FAILURE in 9m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31b13466fa84492090a421a989b7089e : FAILURE in 5m 44s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7f2b62baf13b444d918ad83ada110ffc : FAILURE in 5m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3bcef7c657524cd586db196f0f1c592c : FAILURE in 9m 33s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c8c63e8636b64d868afb0de3dbf7c6fd : FAILURE in 8m 56s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3468c65dda1742c0b0338e6022cf672e : FAILURE in 10m 47s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/90f29bf566614d12980d2e0a96a2dfab : FAILURE in 8m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8317f5e83b6647d48c9bdd31b9e13006 : FAILURE in 10m 33s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b80aa95aa46f4b7d8c5ae8c6717a4a67 : FAILURE in 8m 43s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70dcf21dea8b4cb1ae75ebda8fd3d1a7 : FAILURE in 8m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b63890a42b9446cc83961c03758c4e19 : SUCCESS in 3m 34s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/62d3bf5c005141ce860930208505401e : SUCCESS in 1h 12m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9fb270765f52a3dd1836aeb4f498950f0af3b17d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-06 00:59:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42f2b027780f05243f23e80da1ff8d038f9c476e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 02:17:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e239c8d7fcf487da6ae0893851befc3 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0fd4f65674c489ca5199ac2a3933ca2 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef7c71b9c3194d1ba32fa6642e449428 : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7a832c52b7364abea6ffad5ea25ce0a6 : SUCCESS in 13m 16s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/20e8469632494b92adb215684926a15f : SUCCESS in 13m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/462ceaab832848499b29a43cff70d4dc : SUCCESS in 11m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28d04e17a29b4a35976f2576eb140326 : FAILURE in 11m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f48c4d9200ee472caabe31a5eafc53c0 : SUCCESS in 47m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0a3b4a9f75349c88436d56fc6b113c6 : SUCCESS in 49m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69f196052e3b454292f44d0738f3d337 : SUCCESS in 42m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8787d35bff2546d6b1ff7e80d03a845e : SUCCESS in 38m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ede49b817087459781ec6a6166bcde73 : SUCCESS in 49m 07s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5eb58088a4f94dea9b0d842fb9466262 : FAILURE in 1h 02m 43s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/464bacfb39a949cf9423ca57554aafd3 : SUCCESS in 1h 02m 47s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/76aeedc3bc814d3194f4d19a5c08ebfc : SUCCESS in 44m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a6e7b4a7a692428aa4462e3c56ca2ca8 : SUCCESS in 38m 25s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ac202a70dda84224a6630b7c84d8ffab : FAILURE in 10m 09s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2df5fcfd259d4da9b4393cf3ef21beae : FAILURE in 7m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10e79989414e4f02bf16be78a4499764 : SUCCESS in 3m 50s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a089b233c56f4077b33590cc6c9d0036 : SUCCESS in 1h 15m 47s","accounts_in_message":[],"_revision_number":2},{"id":"83d83e8a8490bf1e755028085ea88ed53c8c3439","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-06 03:30:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"02ee50b0249d43835285cbdcb613bd67485c1711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 04:45:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/781b5df2e2cc4716bd27230eced3af30 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b9d98ac32d64548bc25ecda68fb978f : SUCCESS in 4m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fab1e2de1ef24673976afaec88ed3a95 : SUCCESS in 4m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d2dadceb2c874afaabd2e4af134fe6e8 : SUCCESS in 12m 11s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0079b2531aed4b899837ad43a2bd97e5 : SUCCESS in 13m 09s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c12cc3bbda7543c7841d50f06c8b9655 : SUCCESS in 12m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3558b47424cd4778b929e41b254b988c : FAILURE in 10m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/93684a30d4da4d7b8260200d480d1608 : SUCCESS in 40m 36s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01589d82fd1b4f70b8066895363744f9 : SUCCESS in 50m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78ff6ca4fa824fe6a3cd1dc19fd5091b : SUCCESS in 40m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2c891b9dbbd74a2296e869ef234bab28 : SUCCESS in 36m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/988b1010e29d471580aabc6f5e35f3cb : SUCCESS in 54m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c337480ef2b1423b9f09fab71cefbe08 : SUCCESS in 57m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a1e6f87821104dd6a7db34b04d8fa5eb : SUCCESS in 1h 02m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/624c3cf0b0d94d26b00310cdbbd5c8da : SUCCESS in 55m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e983ac44670b476fbd1efe4c170a57f5 : SUCCESS in 38m 17s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/04c0939c84af4fad8dcc43ec93c5396a : FAILURE in 14m 11s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fa4ac487dbe4f579e060489e85ad568 : FAILURE in 9m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6139f34db7bf4a79b1c5038b340ba674 : SUCCESS in 3m 24s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/01c2f93921ce40398850b1929cc7b104 : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":2},{"id":"61ca673b5d4ab051dff5a54abb394e6a9a97a519","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-06 08:57:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25379ee44896767cb9855a35b2c57f8325182f27","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-06 14:03:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b864dddb9887413098e11cf4bc76154290b346de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 14:03:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09d5246def1fe46cbc354896fdf4e7e50ff167fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:44:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0faa8fddc44a42788cee192beb621cf9 : SUCCESS in 6m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/382547a36b0b45f9b39331dd5f5488b6 : SUCCESS in 5m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/304f80ca795843438be4d0cd954b2eca : SUCCESS in 6m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5a12496bf3e6426c809542ae96148dba : SUCCESS in 15m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/dc9a2b5b6e9340d4bf4229d60e589c4a : SUCCESS in 17m 13s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1f5f8875f9f41f08f5509463407ec8f : SUCCESS in 12m 24s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f6b45067f99d4909a050d08d96eb5bba : SUCCESS in 45m 32s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ca434fba78b842b8a92d9088c8110645 : SUCCESS in 1h 05m 02s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bb4d5fad4fff4e97836d60f3be42bced : SUCCESS in 1h 13m 18s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b1e94379ccda4c47bea1b9a664a69829 : SUCCESS in 44m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ab9c401fc5094690aaaa3d83ac2ea346 : SUCCESS in 46m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7456353a5e1843a29b73f89f69c9f853 : SUCCESS in 5m 00s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/481d65b5396b47db99000ed4be460ab8 : SUCCESS in 1h 26m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6732b6e6e1b27e73d6435fea88f15d10fca8ab98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:44:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"52398d68289dcec941ce8f0462abf89f58542c84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:50:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9f00bd7ca3d4820a0b098590117d7ce : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd8badd39c04ca49ca4c49a6ce7cef26349d3960","revisions":{"f2d7590bc5def9793feba0ef74270f6937f02e51":{"kind":"REWORK","_number":1,"created":"2020-07-05 23:42:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/739380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/80/739380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/80/739380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/80/739380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/80/739380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/80/739380/1"}}},"commit":{"parents":[{"commit":"87499a232a046cfa8f0ac663f435d088ce976533","subject":"Merge \"Prepare Victoria M1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/87499a232a046cfa8f0ac663f435d088ce976533"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-05 23:39:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-05 23:41:57.000000000","tz":540},"subject":"Deprecate keystone::cron::token_fush","message":"Deprecate keystone::cron::token_fush\n\n... because ``keystone-manage token_flush`` was already removed from\nkeystone[1].\n\n[1] https://github.com/openstack/keystone/commit/48dcdb5ad3ea47e923da3ab4a1572159d7cac6df\n\nChange-Id: I8a7abed59468f93d48aa38a47ad4f72d994da79a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/f2d7590bc5def9793feba0ef74270f6937f02e51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/f2d7590bc5def9793feba0ef74270f6937f02e51"}]},"branch":"refs/heads/master"},"fd8badd39c04ca49ca4c49a6ce7cef26349d3960":{"kind":"REWORK","_number":2,"created":"2020-07-06 00:59:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/739380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/80/739380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/80/739380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/80/739380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/80/739380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/80/739380/2"}}},"commit":{"parents":[{"commit":"87499a232a046cfa8f0ac663f435d088ce976533","subject":"Merge \"Prepare Victoria M1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/87499a232a046cfa8f0ac663f435d088ce976533"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-05 23:39:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-06 00:59:39.000000000","tz":540},"subject":"Deprecate keystone::cron::token_fush","message":"Deprecate keystone::cron::token_fush\n\n... because ``keystone-manage token_flush`` was already removed from\nkeystone[1].\n\n[1] https://github.com/openstack/keystone/commit/48dcdb5ad3ea47e923da3ab4a1572159d7cac6df\n\nChange-Id: I8a7abed59468f93d48aa38a47ad4f72d994da79a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fd8badd39c04ca49ca4c49a6ce7cef26349d3960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fd8badd39c04ca49ca4c49a6ce7cef26349d3960"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
