)]}'
{"id":"openstack%2Fpuppet-keystone~740783","triplet_id":"openstack%2Fpuppet-keystone~master~Ife77f742fa875ab32040359b917599ec2ef844c0","project":"openstack/puppet-keystone","branch":"master","hashtags":[],"change_id":"Ife77f742fa875ab32040359b917599ec2ef844c0","subject":"Change to relative includes","status":"MERGED","created":"2020-07-13 19:16:39.000000000","updated":"2020-07-14 00:54:04.000000000","submitted":"2020-07-14 00:54:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740783-1594688044476-6b2a4802","meta_rev_id":"d0f1a98438a906201d94f73a6f5d8705b8d8a392","_number":740783,"virtual_id_number":740783,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 00:54:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-07-13 22:54:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-07-13 22:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-07-13 22:54:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-13 22:03:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-07-13 22:54:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-07-14 00:54:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e0c07a4b0392a32ba6cd454e76e2656cd28f56a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-13 19:16:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e6f2cd2347ffe5d3d62c02f05b6f5b9e05c409f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 21:19:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2837340cf8c4cb6a243f9fe048c6fa7 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90cd3a617a6842fb836198738aba25a2 : SUCCESS in 5m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eefc54fe3bfd4a41b78bb7aa0a1d976b : SUCCESS in 6m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2d5e2566884b4ee097d43fcca56314ab : SUCCESS in 18m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c2fc1deec93f4cc48a5ecb31bf5dda78 : SUCCESS in 16m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa77a59bc225400db216bb9a1f09f167 : SUCCESS in 13m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eda7ef24f1224d1cb3b976e9d7a8faec : FAILURE in 14m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0b083cee770a4ad9be5e2c1dd5ddeb9e : SUCCESS in 52m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8d8948e716148a59c99f562a8e9f9f2 : FAILURE in 57m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9f0ae7415f448f384303082aabd4a49 : SUCCESS in 53m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bb412e82eccf40f29f7214a234ec52ae : SUCCESS in 43m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/93824a91311f44108314df904d76907d : SUCCESS in 1h 50m 58s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dd286ac5ceac4dbfadd71e4dda013614 : SUCCESS in 1h 49m 13s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f04ca4bd582c426c9e78f2e72e3f3471 : SUCCESS in 1h 38m 18s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0725f4c0a5d24a108d4c6264a7ed03b0 : SUCCESS in 56m 50s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5f981a690a1f4fcaae355b8b4c330483 : SUCCESS in 43m 24s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9525012e22a24afab7712613e0b81aaf : FAILURE in 9m 53s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/604299431afc48a2b01ae03ba20f7003 : FAILURE in 8m 53s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c1a286f857f54bce84ece3bfdf493886 : SUCCESS in 1h 36m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c13340cbfeea99d5df93f6690dd0fb09d01b52e3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-07-13 22:03:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd1e5a5e72e86459a9fb9b61f228a8157ee536f0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-13 22:54:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c0a7fb902d60600e1f843200c322a5194be1bbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 22:55:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00faebcb22162f38520d187d7329667d9bdbb390","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 00:54:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d0f1a98438a906201d94f73a6f5d8705b8d8a392","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 00:54:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a49fa16c0e834218ac8c128e4059ab8a : SUCCESS in 4m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b59419d4805477c94c437079ee7088f : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f333f64efce64ae291ca9e4f0e0cfd61 : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c62f256e0dae43fea7971ccc38631abe : SUCCESS in 15m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/be9473ba543b4ee2bcee955c28a52bff : SUCCESS in 14m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/766988e65712449bb6e3ca6ae88b97cd : SUCCESS in 12m 21s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1d6704bfca3f4feeb4cbbcadf2a2897b : SUCCESS in 1h 18m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/adcbbd7f8ea8457988b23b2527b5f300 : SUCCESS in 1h 53m 59s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/60ba9750f2924b5a89331970a5e6bd79 : SUCCESS in 1h 26m 43s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cb1d9de13e734b38b38ab0479aa57fc4 : SUCCESS in 56m 04s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7c370cba69834f39a3fbb58bd1af1d92 : SUCCESS in 41m 32s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/34e5924ce71d4a268dcb05ca46cceb8b : SUCCESS in 1h 41m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fec22dbe8e9424d33f4ae307d267e8e0d2efc8f9","revisions":{"fec22dbe8e9424d33f4ae307d267e8e0d2efc8f9":{"kind":"REWORK","_number":1,"created":"2020-07-13 19:16:39.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/83/740783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/83/740783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/83/740783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/83/740783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/83/740783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/83/740783/1"}}},"commit":{"parents":[{"commit":"4d0889a3dde24f243b56c0798c9f5537443a7aec","subject":"Merge \"Use relative name for class inclusion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/4d0889a3dde24f243b56c0798c9f5537443a7aec"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-07-13 19:16:26.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-07-13 19:16:26.000000000","tz":120},"subject":"Change to relative includes","message":"Change to relative includes\n\nChange-Id: Ife77f742fa875ab32040359b917599ec2ef844c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fec22dbe8e9424d33f4ae307d267e8e0d2efc8f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fec22dbe8e9424d33f4ae307d267e8e0d2efc8f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
