)]}'
{"id":"openstack%2Fpuppet-keystone~743858","triplet_id":"openstack%2Fpuppet-keystone~master~I1ba6aef478592e681887d87380a15ee8667ff006","project":"openstack/puppet-keystone","branch":"master","topic":"service_type","hashtags":[],"change_id":"I1ba6aef478592e681887d87380a15ee8667ff006","subject":"Add support for service_type in authtoken config","status":"MERGED","created":"2020-07-30 00:51:02.000000000","updated":"2020-07-30 16:06:17.000000000","submitted":"2020-07-30 16:03:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743858-1596125026325-c0251aed","meta_rev_id":"fc52793d926c2a95cdd72c7d991ca56bee41371d","_number":743858,"virtual_id_number":743858,"owner":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-30 16:03:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2020-07-30 13:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-07-30 14:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-07-30 14:03:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 13:54:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-07-30 14:03:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-07-30 16:03:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"762d035250172220252660a6ccd98a51f6b4e242","author":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"date":"2020-07-30 00:51:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a61d911c2dccc0c96c289ef6936b113b5890251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 02:44:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36fe8de7d8eb4a84a56f53769d4fdcb9 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/111f47b7769c43d6a46d0b75df99a9d3 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/976639147a35494a80e20050d594eb78 : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/11b0db906fd546e297a764979178fd19 : SUCCESS in 17m 06s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/674d40eba02941c9b169db7d0d90f5de : SUCCESS in 18m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f53ccf401c1f41559132e0430d9703c0 : SUCCESS in 12m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27379fbbefd34ce6968fb298363babdb : FAILURE in 10m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/82d2367a465549bf8e20380a43e15d8e : SUCCESS in 49m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60798fad9d494c3dbae51b2edfcbe7f3 : FAILURE in 1h 00m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e354250eeeb48e98a4ff625807b6c04 : SUCCESS in 54m 16s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8d8c22c4f1af44468be5e5945f55910d : SUCCESS in 39m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1548fb2c39e74d08999011aab78e59c8 : SUCCESS in 1h 17m 28s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d8ce4d1394124c7bb27704c89448ee0c : SUCCESS in 1h 45m 17s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cf089e114bb647199f7d588aecd4c437 : SUCCESS in 1h 21m 32s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cd91ef814f6b44cdabdb2cb2fef681fb : SUCCESS in 59m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/10f512805300447d921970caf83c419c : SUCCESS in 42m 46s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4b8177695ede43f6a3ca76fc76e07a94 : FAILURE in 10m 02s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/303a0b177059475aae959ff2ea87e2d1 : FAILURE in 8m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eae2f3be52d4704b5a1a11a5c5706a9 : SUCCESS in 3m 36s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee508680821c4ef0b5fa6cc81ad80558 : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e7d5addae522cce0091817c91a1ddf31909c9f0c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-07-30 13:54:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29ada09e481b4305d358ad226ec44cbdfdc61864","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-30 14:03:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12c8e8b44f2ac582829bf610dc62039e1759edae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 14:03:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ced9c1720e3dac603a61ee9829888aa67b92a4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 16:03:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"861c403c73f8ce7efdc3329f6550ab7a53df8a50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 16:03:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b13968f6a33a4f84842a81da88ec55f7 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/008d9a7d4d5c409297323edb0c78c8e0 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e2008b0abca43ecaf2a9e7637f9a0f5 : SUCCESS in 4m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1a9e666a6ffe492096e86f68ac67c038 : SUCCESS in 13m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1c0be310b0ab4fd4ac16e6cd05bb09c1 : SUCCESS in 15m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9ddd6d7d5e04fe89a8cce53a638521c : SUCCESS in 11m 08s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aac8397ed4074a268f0ab6351d0f5689 : SUCCESS in 1h 23m 22s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b1f1ccfb34b64cc19f8110296a5dc1a9 : SUCCESS in 1h 21m 59s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4b1496a5974240af95608bf46da91f00 : SUCCESS in 1h 37m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f1b3ebc05e704bfdbd42744197640738 : SUCCESS in 55m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/66efc23b589b4da6951df48164deba69 : SUCCESS in 41m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a0d10e223d1419e8b5db146ce277548 : SUCCESS in 4m 10s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c621a71bc01246afa0e57a7f56a89cd0 : SUCCESS in 1h 27m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4ed527d24c46d911fe5d8372d95c4d71405f69e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 16:06:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62ff5b6ba00e42798e0c4134ce7790b2 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aba4a00d2998f4b648fa8bed9947c8d5413884bd","revisions":{"aba4a00d2998f4b648fa8bed9947c8d5413884bd":{"kind":"REWORK","_number":1,"created":"2020-07-30 00:51:02.000000000","uploader":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"ref":"refs/changes/58/743858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/58/743858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/58/743858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/58/743858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/58/743858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/58/743858/1"}}},"commit":{"parents":[{"commit":"d5e3a563e65a7d6670fdfbceca3e8650b9cdd9d9","subject":"Merge \"Convert more to rspec-puppet-facts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d5e3a563e65a7d6670fdfbceca3e8650b9cdd9d9"}]}],"author":{"name":"Kieran Spear","email":"kispear@gmail.com","date":"2020-07-28 06:41:45.000000000","tz":600},"committer":{"name":"Kieran Spear","email":"kispear@gmail.com","date":"2020-07-30 00:49:52.000000000","tz":600},"subject":"Add support for service_type in authtoken config","message":"Add support for service_type in authtoken config\n\nThis is used to validate tokens that have restricted access rules.\n\nChange-Id: I1ba6aef478592e681887d87380a15ee8667ff006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/aba4a00d2998f4b648fa8bed9947c8d5413884bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/aba4a00d2998f4b648fa8bed9947c8d5413884bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
