)]}'
{"id":"openstack%2Fpuppet-keystone~746646","triplet_id":"openstack%2Fpuppet-keystone~master~Ia1208fc902064ce5cf3dc769e72cade497142e68","project":"openstack/puppet-keystone","branch":"master","topic":"746643","hashtags":[],"change_id":"Ia1208fc902064ce5cf3dc769e72cade497142e68","subject":"Remove deprecated database_min_pool_size parameter","status":"MERGED","created":"2020-08-18 07:24:09.000000000","updated":"2020-08-24 17:47:26.000000000","submitted":"2020-08-24 17:45:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746646-1598291128275-1545f216","meta_rev_id":"190e8c527d1ebc47ef7363bf73ad9d060adae841","_number":746646,"virtual_id_number":746646,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-08-21 21:39:26.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-24 17:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-24 15:51:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-22 12:31:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-24 15:51:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-22 12:31:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-08-24 15:51:29.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-24 17:45:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1646efc177376690e9f8fe25d389e183b78523fe","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-18 07:24:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04da13d8742e92b2152c35070f1fc4b3f25c8dec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 09:08:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52743909a6c8499e9ecd6ba08b7ba5c9 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9b3516190374d2eb48e8a015fe9b458 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18c3612001b54a8b93d4ac12f6b1633f : SUCCESS in 5m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7bd5336f9a014c61a9116c240e5474fc : SUCCESS in 16m 05s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/aaf49fcd9d36489293b122e2aef4ada8 : SUCCESS in 17m 07s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7696f8436e01428e9eaf0c22ae4afc52 : SUCCESS in 14m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a75911bfc71e44c5bcc701f860412c30 : FAILURE in 11m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/33de99a064b745f3931df5b2d139329b : SUCCESS in 55m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10428416abba40efab54676d7afc06ce : FAILURE in 52m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bedd5461c304f7e85ed2d6faa909051 : SUCCESS in 53m 22s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f72f40ea1b6f4ce0853754e5defeb3cc : SUCCESS in 50m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/712dd5b0575d4f6d9317d3f06d7d6bb4 : SUCCESS in 1h 08m 50s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7cd7f8343f7b4359890b51ebe1ce4a1d : SUCCESS in 1h 21m 04s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d1619e38fd8b410aba893eacb3b736d2 : SUCCESS in 1h 24m 07s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dbbd42fea9fc4e7994e684360f148e9d : SUCCESS in 51m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/089eaee172ec433c9d5ed5ed9042be75 : SUCCESS in 43m 06s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f233b769357644789cfb034b3fbf0134 : FAILURE in 9m 12s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/195db77ad0344eb5aca6759275fa55e8 : FAILURE in 7m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c2dd6c7ce5440ac82408ffa7339ef95 : SUCCESS in 3m 52s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/15f8b6eef9274d09968d1a9de29be2c6 : FAILURE in 51m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c3f6461b6145c1bb0d7a4963a2a036c51a2dcd33","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-21 08:03:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4ee76350ba2f0d4b4fce49deecaf33c9e19c305f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:29:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/344a429f80cd4a9fa12e0a995dc78888 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5efd4eb4a916465994a95a8f866cacca : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77d7a99aaf5c40958d0424e686556966 : SUCCESS in 5m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4e2347ec346945b09c9f5db5eeaa3b10 : SUCCESS in 15m 11s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1854624f343e487daccb37dfe1072c52 : SUCCESS in 15m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85c1e724887c4e95a3de9b854af90d05 : SUCCESS in 14m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1cbe915b4534110aa074401a5a9637f : FAILURE in 9m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cff40e8d575e49b6ba1fde2b8fbb872d : SUCCESS in 50m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c79a560161d44d7a3bfd5dae4ea7fd9 : FAILURE in 53m 23s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eae5c553dc7748e794f1019c2d444fce : SUCCESS in 43m 15s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5d3109b8491e4db4aeb405cfcfdfc0b4 : SUCCESS in 38m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8b8eedf483544630bb7167b18e2cca60 : SUCCESS in 47m 45s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/926f6038285f458b839b68c278d965ec : SUCCESS in 1h 34m 41s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fc524515324e42a19df72fee3e092622 : SUCCESS in 1h 24m 01s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7c43b42296c745d5a05f1221294bf647 : SUCCESS in 57m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e9a577e4b25544e0a8f0dfcf481e944c : SUCCESS in 38m 57s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/5e64d879373c40019a2557557ed172bd : FAILURE in 9m 58s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a41a311246c143108e048fbffdabda56 : FAILURE in 8m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10eeaf1745fd4da9a45b7d56bfb0b342 : SUCCESS in 4m 37s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/12bad77c57704e3080d060d3ad5f5da4 : FAILURE in 1h 16m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cea7e17ffd64214ad2c9354871bd6439d18db40f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-21 21:39:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a4de43cf7348a7506e2f0faa27588eab4e7edfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 23:24:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d11125b5197a4b5e83364555cc6c1dde : SUCCESS in 4m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9f94245f35e42689bfdc51d9ed05bf3 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a49118857bcf4078a28356efa69425a7 : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4e074006b7ef494e96a52a6925d70bc9 : SUCCESS in 14m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/adcf03ea49034ac9a0e2a5d84fb6631d : SUCCESS in 15m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc3ca5f9ba2e4a4294211f1c589203e1 : SUCCESS in 14m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/618b68b615ee48679fb81c5c5678a0e0 : FAILURE in 6m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/305e2c17456f4e549a47e5dc96b4e981 : SUCCESS in 44m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/532c6b8a68bb44a2935ec4b46f1fd05f : FAILURE in 50m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28cf15893bb843ed9867ecd5f8404be6 : SUCCESS in 43m 02s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8aa27cb83ee143d4b6f31bfced5a5a40 : SUCCESS in 43m 22s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/47c7110723294d51beae4647d86e0f56 : SUCCESS in 1h 16m 48s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2df4c2974c05489d9ea1415a870ffe53 : SUCCESS in 1h 29m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f328819272a64300b1ec6089a582163b : SUCCESS in 1h 29m 53s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e8604638688443999d7cced02282e667 : SUCCESS in 55m 33s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7c8945beca684033bb0c7c933c3dc563 : SUCCESS in 40m 30s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0fe977f00b8e49f5ae751dd5ab66175e : FAILURE in 10m 04s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99b09abee0b3403c8f21e7429b876bd6 : FAILURE in 7m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31816028d5ca47e39ed777df93b96086 : SUCCESS in 4m 45s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/16b93ba3549248eb9b9691553aee0449 : SUCCESS in 1h 34m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8df7f13dcea7da71fc5082b958400b20008143a5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-22 12:31:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c6b45c70cf704f4c3707d586717e4df1db3d154","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-24 15:51:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95cf26489cbf38ff3a03d3dd6fefbd5f11ef2718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:52:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8f0ae9484595b62deaf4fd19d8aaaada25b5330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 17:45:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8557e0edcf44e531c1c0f53be259b15637eb0f80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 17:45:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bffc28aeb64b4d0e85e5c41ddf208778 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e689b788258d452d854b98e8f505ea8a : SUCCESS in 4m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45350d022a404f0985392ab1cb75875c : SUCCESS in 5m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/122d0c4fd20541178df9899fd90cec15 : SUCCESS in 10m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2bb27bc511024a3d8e4b252243011388 : SUCCESS in 12m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b3d3e7f8db4446aacd7fd8b8d090797 : SUCCESS in 12m 21s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/27d040e5956b4e55a44f687daffae447 : SUCCESS in 1h 19m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/24b3db0d080e47fca2132f1cd66e09d6 : SUCCESS in 1h 43m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0d8a840b6d094c20bec6346c7e4e4f37 : SUCCESS in 1h 29m 21s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4ac09000300d4a92bbd429b5015618a4 : SUCCESS in 1h 02m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9bf7e918f23f468c8ca589edd3c3b76e : SUCCESS in 32m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8498501e2c64014abc1b92e0a80603a : SUCCESS in 4m 46s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/190a216e00d440f085d7b8874106e8ab : SUCCESS in 1h 46m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3aaf00000f31cc73947b2d8b2eb518b29d3b3769","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 17:47:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca802f8a44994e149a36194da897b4c1 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35f5e5e5201c28ff4624994f44475ad269edfb0e","revisions":{"35f5e5e5201c28ff4624994f44475ad269edfb0e":{"kind":"REWORK","_number":1,"created":"2020-08-18 07:24:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/746646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/46/746646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/46/746646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/46/746646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/46/746646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/46/746646/1"}}},"commit":{"parents":[{"commit":"ee755ad4ed14ae11a871673acaf5063bc3dc64de","subject":"Remove deprecated validate parameters from init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/ee755ad4ed14ae11a871673acaf5063bc3dc64de"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-18 07:23:56.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-18 07:23:56.000000000","tz":120},"subject":"Remove deprecated database_min_pool_size parameter","message":"Remove deprecated database_min_pool_size parameter\n\nChange-Id: Ia1208fc902064ce5cf3dc769e72cade497142e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/35f5e5e5201c28ff4624994f44475ad269edfb0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/35f5e5e5201c28ff4624994f44475ad269edfb0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
