)]}'
{"id":"openstack%2Fpuppet-keystone~748953","triplet_id":"openstack%2Fpuppet-keystone~stable%2Ftrain~I0dc5dd487996307811edd177d13d6d5ac8f841ee","project":"openstack/puppet-keystone","branch":"stable/train","topic":"py3-packages","hashtags":[],"change_id":"I0dc5dd487996307811edd177d13d6d5ac8f841ee","subject":"Fix python package names (ldappool and pysaml2)","status":"MERGED","created":"2020-08-31 05:35:12.000000000","updated":"2020-09-01 14:04:53.000000000","submitted":"2020-09-01 14:04:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748953-1598969093961-46488824","meta_rev_id":"fc46f6fd46ddad902226ce7feef71f6dbe0aaacd","_number":748953,"virtual_id_number":748953,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 14:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-31 13:27:54.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-31 13:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-31 13:27:54.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 13:13:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-31 13:27:54.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-01 14:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4379868c2fd274e6e9a200f9828f295bf9f2d79b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 05:35:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfb86748cec1f7fb3e02afed67d932a22558a1de","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 05:38:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1da265be3c0396ab88e5f26a428c873854d0c16b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 06:48:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c6f63288ea04364b60d56ef3f870da8 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d4be40ad48846f4b51f9cd3d3bf10c0 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cacd801424a441f78a26c78d6df41b91 : SUCCESS in 4m 42s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/550326ce0faf46adaaf3876233182025 : FAILURE in 19m 16s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4cc2eb5078ce41a697f34916639365e4 : FAILURE in 16m 05s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/49da76c5e7434c3885266e78f21b7f8a : FAILURE in 16m 52s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e1acd4aaf50498b899fe5f6e8a63a00 : FAILURE in 11m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6425ac1e28ef494e8fc0f9e0f189c0c4 : FAILURE in 17m 50s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c84ae68e5d6249f6bf004ce61b1bde7d : SUCCESS in 1h 01m 22s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f07cd0f838b247f68bf35b8144aae37a : SUCCESS in 1h 05m 52s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fb6deffbd5bb45cda7f7424d00f651a3 : SUCCESS in 1h 07m 03s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/d7dcfd167fe14d17b32d23bd3e58db66 : SUCCESS in 55m 27s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4250a1c4bcd64ebb83ce6c08f7331ae8 : SUCCESS in 55m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47c531ea89164792b4d54a46aefe7ea2 : SUCCESS in 59m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf2275637eb54532b98962594f13a79f : SUCCESS in 49m 00s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/301651201dbe4a85bb9dcea8e4dd2d7a : SUCCESS in 49m 41s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9636bd9f1b7d40c5a73a81a1b7646484 : FAILURE in 11m 21s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5dd40d889c1b4cb1af65574735523339 : FAILURE in 11m 19s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f1bec2a2d0414aabb9655ab936311428 : FAILURE in 11m 02s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3ec446cd697f4d9383201a361122c9ae : SUCCESS in 51m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/da4a32658c83409991286ff14733d1cf : SUCCESS in 56m 48s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22e47bb1152d46dd939d9b1b58deae6a : FAILURE in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"328802722a0c3aa1f95f1b579205c2442f36f08f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 09:27:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"880bf1ed62a000dca4546957224599d26f067e32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 10:37:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1a4f34931614102a953afa7b2e3be8e : SUCCESS in 4m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41a4784cdf1840a29b749581b7f068e2 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7faa9a67b9d495091294f3f67da0231 : SUCCESS in 4m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4b635ce7d5654ce0984b10c7968a98bb : SUCCESS in 17m 06s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8799e922e90544eca914ecf6b613b7c3 : SUCCESS in 11m 04s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/ce5f78e24c2e47428df1afd29e24f887 : SUCCESS in 14m 43s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be124327ffb4427a801fd6ab6e54144f : SUCCESS in 8m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ccdba9c0c7b44b4b2937bbc1fd4d4bd : FAILURE in 17m 06s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/45c5895d0e0649b694a973bdcf815a6a : SUCCESS in 50m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/191c09a6750e4e9db297f7fb0c9ed7ea : SUCCESS in 1h 02m 31s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/569e95efd451492a9aafd338e144ee28 : SUCCESS in 1h 06m 38s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/749066d916e44841bdd6998fc8479287 : SUCCESS in 52m 38s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1279339763cc443d8ef4c3ed5439981c : SUCCESS in 50m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f248f9ef0164b458508bdfeb97826fd : SUCCESS in 54m 27s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b55cf0bda8d84c09a60c0688132dde51 : SUCCESS in 44m 34s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0d53e7a2fb884ac0a0b6907a57fa68d5 : SUCCESS in 34m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b649a6beea334ee6b88ba3304690631b : FAILURE in 25m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/91b9f1a112b14e87a0787121b6a0455d : FAILURE in 23m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0d4a2ae569964d5380a311b21aa2a81f : FAILURE in 9m 03s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d477ee5f23914c58a360ae89c8015848 : SUCCESS in 43m 29s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c458fc9b61c647caae0a653c131761b5 : SUCCESS in 39m 07s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d7cf6e8e8e5456faf0ad14ca4caa42c : FAILURE in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7b0d26bfe353f2e365a685f34483ed31a3131bc6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 13:13:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0fd6faea8099b25a6d56be5e98dbade1185eef1f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-31 13:27:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a041eb87369a1ac813860672b12417f41a7a2c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 12:51:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a7746534b3d65c0dacd0da1263c150fa3ed1f2b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:04:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/875a8b15e30d4a1e92ba3815b9f586ef : SUCCESS in 5m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c893831f7a5d4826a53e95ae340fff98 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3507365a9ef8443ea3b13fcf13c522a7 : SUCCESS in 5m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/eb990bd836ab42729bc6b2016c4c1b17 : SUCCESS in 26m 45s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bfb9903e42a4191baa593d05ff5569b : SUCCESS in 14m 59s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/eef2783fff8b43aaa28fe441eb671a77 : SUCCESS in 57m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/21205d384368445c98e5d476cb6bb52b : SUCCESS in 57m 03s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2f0a313702d1468295fe180f1b7a8a15 : SUCCESS in 1h 00m 52s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/07dad8034ff84a30807f057972059368 : SUCCESS in 55m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2d5fb35331fb4b4b847b972f7f054814 : SUCCESS in 50m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0cc29a18f440b6403e8907efc048790856830719","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:04:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1660e2c8de6860e16e5a567b7e36e4e54fcb8a49","revisions":{"83351118633fdb29227b76e8d94c1002489f5a78":{"kind":"REWORK","_number":1,"created":"2020-08-31 05:35:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/748953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/53/748953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/1"}}},"commit":{"parents":[{"commit":"1e38a1c578c3201f3d9c377bdd8acc219aad766d","subject":"Convert more to rspec-puppet-facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1e38a1c578c3201f3d9c377bdd8acc219aad766d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 12:58:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 05:34:47.000000000","tz":540},"subject":"Fix python package names (ldappool and pysaml2)","message":"Fix python package names (ldappool and pysaml2)\n\nThis patch fixes package name of python-ldapppool and python-pysaml2,\nso that correct package names, python3-*, are used in OSs which use\nPython3 as the default Python.\n\nConflicts:\n\tmanifests/ldap_backend.pp\n\tspec/defines/keystone_ldap_backend_spec.rb\n\nChange-Id: I0dc5dd487996307811edd177d13d6d5ac8f841ee\n(cherry picked from commit f342aad94615655fe96959186c031157c209a659)\n(cherry picked from commit 2917cc9262a3187d0e01b039efa744ab7f2943ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/83351118633fdb29227b76e8d94c1002489f5a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/83351118633fdb29227b76e8d94c1002489f5a78"}]},"branch":"refs/heads/stable/train"},"620a785860b93d56f05b6c0f488997ab95980f87":{"kind":"REWORK","_number":2,"created":"2020-08-31 05:38:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/748953/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/53/748953/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/2"}}},"commit":{"parents":[{"commit":"1e38a1c578c3201f3d9c377bdd8acc219aad766d","subject":"Convert more to rspec-puppet-facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1e38a1c578c3201f3d9c377bdd8acc219aad766d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 12:58:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 05:38:09.000000000","tz":540},"subject":"Fix python package names (ldappool and pysaml2)","message":"Fix python package names (ldappool and pysaml2)\n\nThis patch fixes package name of python-ldapppool and python-pysaml2,\nso that correct package names, python3-*, are used in OSs which use\nPython3 as the default Python.\n\nConflicts:\n\tmanifests/ldap_backend.pp\n\tspec/defines/keystone_ldap_backend_spec.rb\n\nChange-Id: I0dc5dd487996307811edd177d13d6d5ac8f841ee\n(cherry picked from commit f342aad94615655fe96959186c031157c209a659)\n(cherry picked from commit 2917cc9262a3187d0e01b039efa744ab7f2943ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/620a785860b93d56f05b6c0f488997ab95980f87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/620a785860b93d56f05b6c0f488997ab95980f87"}]},"branch":"refs/heads/stable/train"},"1660e2c8de6860e16e5a567b7e36e4e54fcb8a49":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-31 09:27:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/748953/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/53/748953/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/53/748953/3"}}},"commit":{"parents":[{"commit":"a34e31d0f71bfab54c02c5209534b6ab8ac0ef50","subject":"Convert more to rspec-puppet-facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/a34e31d0f71bfab54c02c5209534b6ab8ac0ef50"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 12:58:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 09:27:56.000000000","tz":0},"subject":"Fix python package names (ldappool and pysaml2)","message":"Fix python package names (ldappool and pysaml2)\n\nThis patch fixes package name of python-ldapppool and python-pysaml2,\nso that correct package names, python3-*, are used in OSs which use\nPython3 as the default Python.\n\nConflicts:\n\tmanifests/ldap_backend.pp\n\tspec/defines/keystone_ldap_backend_spec.rb\n\nChange-Id: I0dc5dd487996307811edd177d13d6d5ac8f841ee\n(cherry picked from commit f342aad94615655fe96959186c031157c209a659)\n(cherry picked from commit 2917cc9262a3187d0e01b039efa744ab7f2943ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1660e2c8de6860e16e5a567b7e36e4e54fcb8a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1660e2c8de6860e16e5a567b7e36e4e54fcb8a49"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
