)]}'
{"id":"openstack%2Fpuppet-keystone~760383","triplet_id":"openstack%2Fpuppet-keystone~master~I78c35f92370ca5a7cd213d9b82308366c8de0446","project":"openstack/puppet-keystone","branch":"master","topic":"openidc.pp_UPDATE","hashtags":[],"change_id":"I78c35f92370ca5a7cd213d9b82308366c8de0446","subject":"DNM: Make sure OIDCMemCacheServers is not set by default","status":"ABANDONED","created":"2020-10-29 15:01:56.000000000","updated":"2021-08-17 14:00:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f6742ae1244f3e86a9559a3d4346f40544e0db61","_number":760383,"virtual_id_number":760383,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 15:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ae504e3330dbc9e87ca40b4675f9da75cf8c721","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-29 15:01:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c7a606119231d0ec1bf3685f99cba8d96aaae5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 15:22:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80938f1474b94d5ba57093754a745e5c : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/084039576e6c46eb93223c1f7bbcd7a2 : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5f44bc4b546449c784467731f02fa9ce : FAILURE in 13m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db4cce5957c74c1f91b00dcab228fd9f : FAILURE in 9m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59c4a12e330642eb81a6889f07337d34 : FAILURE in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0142eb58cc6ab2d6dc2ac151ceca6b73381a8e7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-20 14:47:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f6742ae1244f3e86a9559a3d4346f40544e0db61","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-17 14:00:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"756e38aa1943a2626818349c9651e8e25125eadb","revisions":{"e81f9a140adec094f99b90bf694696f46adb6da4":{"kind":"REWORK","_number":1,"created":"2020-10-29 15:01:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/760383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/83/760383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/83/760383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/83/760383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/83/760383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/83/760383/1"}}},"commit":{"parents":[{"commit":"5c1632c100e89437d17ddae40ee9723fcb9e6c67","subject":"Unset memcached_servers by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5c1632c100e89437d17ddae40ee9723fcb9e6c67"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-29 15:01:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-29 15:01:29.000000000","tz":540},"subject":"DNM: Make sure OIDCMemCacheServers is not set by default","message":"DNM: Make sure OIDCMemCacheServers is not set by default\n\nChange-Id: I78c35f92370ca5a7cd213d9b82308366c8de0446\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/e81f9a140adec094f99b90bf694696f46adb6da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/e81f9a140adec094f99b90bf694696f46adb6da4"}]},"branch":"refs/heads/master"},"756e38aa1943a2626818349c9651e8e25125eadb":{"kind":"REWORK","_number":2,"created":"2020-11-20 14:47:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/760383/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/83/760383/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/83/760383/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/83/760383/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/83/760383/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/83/760383/2"}}},"commit":{"parents":[{"commit":"5c1632c100e89437d17ddae40ee9723fcb9e6c67","subject":"Unset memcached_servers by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5c1632c100e89437d17ddae40ee9723fcb9e6c67"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-29 15:01:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-20 14:47:45.000000000","tz":0},"subject":"DNM: Make sure OIDCMemCacheServers is not set by default","message":"DNM: Make sure OIDCMemCacheServers is not set by default\n\nChange-Id: I78c35f92370ca5a7cd213d9b82308366c8de0446\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/756e38aa1943a2626818349c9651e8e25125eadb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/756e38aa1943a2626818349c9651e8e25125eadb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
