)]}'
{"id":"openstack%2Fpuppet-keystone~768725","triplet_id":"openstack%2Fpuppet-keystone~master~I6effcf58a254b6d46ac48a3eb40bfb064b4d851f","project":"openstack/puppet-keystone","branch":"master","topic":"refactor-init","hashtags":[],"change_id":"I6effcf58a254b6d46ac48a3eb40bfb064b4d851f","subject":"Create an independent class for credential feature","status":"ABANDONED","created":"2020-12-29 13:50:03.000000000","updated":"2022-01-06 13:27:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f2572b248df1dc3acdc9f427f832cf243eca791f","_number":768725,"virtual_id_number":768725,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-29 16:40:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 16:40:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2eb8064999f2e67dc5ad964a9d6538118a78a87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-29 13:50:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a997a383aed0e4c270c48e324d98c064910658c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-29 14:29:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed51b441dffbc265a5f194793a07945c52a86166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-29 14:34:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a72b95cace6cdd587ec5081bc24b42a6fbf3edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 16:40:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56d7a5291ccf4e29b6feeb3e2021a177 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e73ebe62dda42ff83bbd7828fdcd361 : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4889183fca93428c82974651f15477dd : FAILURE in 14m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06383c66e700434daf8ee1d39bb1cd87 : FAILURE in 12m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd65989397a247d5971d5b7d94555dc6 : FAILURE in 11m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a67673dc5fb04f11b800204f0371fcfd : SUCCESS in 50m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf1238802850419cb3bb6656ff177a9a : SUCCESS in 50m 54s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99994a17bdb7494d836984f4d5833423 : SUCCESS in 55m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2758143ca99a44c5862b899088d1bb6a : SUCCESS in 39m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c4e56839e5c64e0dbcba92521efca17d : SUCCESS in 1h 30m 24s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ac5a314ae242476ab0a2052f8423fd49 : SUCCESS in 1h 12m 19s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/65619d15923c4e05966991bf783bdada : SUCCESS in 1h 17m 35s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/98bea2b91f974842ba0997e7b1a03b01 : SUCCESS in 1h 15m 04s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/717938ba864c4af6821f8cede2e45105 : SUCCESS in 37m 15s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1f57b20ddb04137b2e8125c34980d1e : SUCCESS in 33m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/758aad647e7e4bf29aa7fdf34cb96c1b : SUCCESS in 3m 51s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a4541302bed74f2cad25441979074c86 : SUCCESS in 2h 01m 33s","accounts_in_message":[],"_revision_number":3},{"id":"f2572b248df1dc3acdc9f427f832cf243eca791f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-06 13:27:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"73e8b828063740f012d8ec9c148cbf59e926945e","revisions":{"0aba76801f44ebde126c6837ba582bdbb8f54d67":{"kind":"REWORK","_number":1,"created":"2020-12-29 13:50:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/768725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/25/768725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/1"}}},"commit":{"parents":[{"commit":"56c39a7370e475f0540195ec1119c50128ef359f","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/56c39a7370e475f0540195ec1119c50128ef359f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-29 13:49:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-29 13:49:02.000000000","tz":540},"subject":"Add keystone::credentials class to configure the credential feature","message":"Add keystone::credentials class to configure the credential feature\n\nThis patch intorduces the independent keystone::credential class to\nmaintain configurations for the credential feature, to make the base\nkeystone class simpler.\n\nChange-Id: I6effcf58a254b6d46ac48a3eb40bfb064b4d851f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0aba76801f44ebde126c6837ba582bdbb8f54d67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0aba76801f44ebde126c6837ba582bdbb8f54d67"}]},"branch":"refs/heads/master"},"fb972885e7e2c1abd304de4830f99bdd45d26bbf":{"kind":"REWORK","_number":2,"created":"2020-12-29 14:29:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/768725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/25/768725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/2"}}},"commit":{"parents":[{"commit":"56c39a7370e475f0540195ec1119c50128ef359f","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/56c39a7370e475f0540195ec1119c50128ef359f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-29 13:49:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-29 14:28:18.000000000","tz":540},"subject":"Create independent class for credential feature","message":"Create independent class for credential feature\n\n... to make the base keystone class simpler.\n\nChange-Id: I6effcf58a254b6d46ac48a3eb40bfb064b4d851f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fb972885e7e2c1abd304de4830f99bdd45d26bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fb972885e7e2c1abd304de4830f99bdd45d26bbf"}]},"branch":"refs/heads/master"},"73e8b828063740f012d8ec9c148cbf59e926945e":{"kind":"REWORK","_number":3,"created":"2020-12-29 14:34:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/768725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/25/768725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/25/768725/3"}}},"commit":{"parents":[{"commit":"56c39a7370e475f0540195ec1119c50128ef359f","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/56c39a7370e475f0540195ec1119c50128ef359f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-29 13:49:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-29 14:31:35.000000000","tz":540},"subject":"Create an independent class for credential feature","message":"Create an independent class for credential feature\n\n... to reduce complexity of the base keystone class.\n\nChange-Id: I6effcf58a254b6d46ac48a3eb40bfb064b4d851f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/73e8b828063740f012d8ec9c148cbf59e926945e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/73e8b828063740f012d8ec9c148cbf59e926945e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
