)]}'
{"id":"openstack%2Fpuppet-keystone~781429","triplet_id":"openstack%2Fpuppet-keystone~master~Iae9098ac8a42c8371a817993d4450747a51ff6f8","project":"openstack/puppet-keystone","branch":"master","topic":"enforce_new_defaults","hashtags":[],"change_id":"Iae9098ac8a42c8371a817993d4450747a51ff6f8","subject":"Add support for oslo_policy/enforce_new_defaults","status":"MERGED","created":"2021-03-18 11:57:57.000000000","updated":"2021-03-22 18:59:23.000000000","submitted":"2021-03-22 18:59:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781429-enforce_new_defaults","meta_rev_id":"ddb0f3fa7aeba6dd8b0fae71704169bb9a49eaa3","_number":781429,"virtual_id_number":781429,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 18:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-03-22 08:22:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-03-22 16:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-03-22 16:49:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 14:17:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 08:22:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-22 16:49:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"7a7a83ce35ecd2851164eda8605b9bc4df3c857a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 11:57:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12af20da614f219c342c5b08c5df11e623fa8677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 11:59:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5fb3c385681854fa1523692827e3b60da518299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:17:44.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"9935285f7b97115fe1ae1c857b122ecbed8cb66d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-19 11:58:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"03ec883565dfdffbdba6de00f8e1dd1e2ded3fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 14:01:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e75d44145bed4134ba7c0d188f7c6c5f : SUCCESS in 5m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9190f455f954404a4f50f5f5e33c12b : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c40285e6baa24a55af570da3481776a9 : SUCCESS in 13m 42s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bae773c32ec46cd983c0d0ec196eb73 : SUCCESS in 14m 40s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86ba5e9d09e144dfa6f12dd846d27c69 : SUCCESS in 11m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8465289efa7b433ab8f8b8fa1c2c205f : SUCCESS in 11m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bbbf4fd63a14073b0e3287aed3594be : SUCCESS in 56m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1f7a830ff1046f8a32f4666e70d5ba0 : SUCCESS in 48m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1336d9d821be42ed9293adb244149e15 : SUCCESS in 48m 16s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/451e09872bca4fc09ba769327e977b83 : SUCCESS in 39m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4b09f362cd949399a721da7f214da35 : SUCCESS in 1h 30m 53s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0aa58f7a04764f9a8f78c611c3c4ea26 : TIMED_OUT in 2h 02m 08s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/90f8f39b17644b3ebe78844e06235ff7 : SUCCESS in 1h 29m 21s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1cf586fb3f424a0c8f5bb36230f2d1fc : SUCCESS in 1h 09m 29s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/be4b4e94efc4427c9af66833096b918e : SUCCESS in 1h 11m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f2f689aba734459b1fe3efb41575768 : SUCCESS in 1h 34m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f87ba006e4b748fd9a54d0ddbd618161 : SUCCESS in 1h 35m 32s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/43160d5725974521b2fe36a262f980b1 : SUCCESS in 1h 05m 31s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/87ddb316db8e4cd693b0f1e2a598d43d : SUCCESS in 42m 39s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a53dcd0e982949a69dcbd8d64e0274d8 : SUCCESS in 47m 13s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/800db9cced4e4e7984c078af9eec62aa : SUCCESS in 34m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4c3b0a319704073966a6050ab902f38 : SUCCESS in 4m 36s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5410b33638f34fe39f196d821ebab992 : SUCCESS in 1h 40m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c3874ba6c545c2249aabd9b1b19b1945ea9c3e34","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-20 12:17:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"54eb36e521f16ba89b09165c97058a78d725c70b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 14:29:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ddc3a2d5d8d4344bcf3f9df84cce52a : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4606f367876345b2b150e340236b68d4 : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/19550d6a17d74000a60fe1ba5787a0ea : SUCCESS in 12m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38b3d4c633ac4bc8975640a02da5f0c0 : SUCCESS in 8m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6d68489b8714f87aee7a5dd8674b10f : SUCCESS in 11m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b7d85732bd34fcfbb76648d27b4242c : SUCCESS in 55m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f26cf15ea60c4fdeaadb9d6d0d002e44 : SUCCESS in 53m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d4865bafd7049779838de73eb85af61 : SUCCESS in 43m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33fc6bea5171461f8e9db5105b5bf0fe : SUCCESS in 35m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/02f3d38409ed4add8392b1fa79569b7e : SUCCESS in 1h 11m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db1e31f816984c5d91826c98e462d4e5 : SUCCESS in 1h 20m 06s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/41569cae392948c5b60184177aeaf65f : SUCCESS in 1h 17m 59s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f34e7e805b4449aa98d12466a717740 : SUCCESS in 56m 44s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d010504440ec4334aef3b8263bdd50c7 : SUCCESS in 45m 21s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77cf45c8d0d6473eb437355a963ae2fe : SUCCESS in 36m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87fa6805492f40e8a8cb8b6a2dc48375 : SUCCESS in 4m 41s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9d43a8ecc12c43eeb04ae27960f70c9b : SUCCESS in 2h 06m 40s","accounts_in_message":[],"_revision_number":2},{"id":"bc7abe5b539374ea10ddd2fe49379f147d106d51","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-22 08:22:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aa6609426a45e039a2eb212495e5665e7e336218","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-22 16:49:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8839e027c0379d803d11a1240d1e0fc95ccce2bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 16:51:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a51cdb3bf1a390ada5ec29b7dd758693258382d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 18:59:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb9a78dd99f24d6b9700881d70dd5f90 : SUCCESS in 5m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/288cb47035ef41cca41467b0d8643d70 : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcdb0b46e4204d2897514b7d3eb4a4f0 : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc5786c72886460da072f9c1f1edeb1e : SUCCESS in 10m 10s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc88ac2e9e474d06ba3f03eb9621f697 : SUCCESS in 1h 10m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/594fd974780a4140b8cf292e03c64d4f : SUCCESS in 1h 28m 20s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4abf82cd23f642c89cb2a6d0f474a5db : SUCCESS in 1h 19m 36s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/483683f996814b3d9ec406692bedc6f0 : SUCCESS in 1h 11m 28s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f07e042bada476eb7da6847b4fca461 : SUCCESS in 33m 49s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2637e701d5440a78e95a42c1e92410f : SUCCESS in 43m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb8f33ad30f645c5a8b00f0c3547daf4 : SUCCESS in 6m 13s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eb284f47ce4f4f1495548698a593844c : SUCCESS in 2h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ddb0f3fa7aeba6dd8b0fae71704169bb9a49eaa3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 18:59:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ba0efa921e411ea309d22bf90d9b9535f9eea37","revisions":{"b2a16e4460bf98a5ff750cd8bafb4804e0450a0c":{"kind":"REWORK","_number":1,"created":"2021-03-18 11:57:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/781429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/29/781429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/29/781429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/29/781429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/29/781429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/29/781429/1"}}},"commit":{"parents":[{"commit":"351d69df2d5230639083c6b576aa58dfe10c36c2","subject":"Prepare Wallaby M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/351d69df2d5230639083c6b576aa58dfe10c36c2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:57:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:57:27.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nDepends-on: https://review.opendev.org/781428\nChange-Id: Iae9098ac8a42c8371a817993d4450747a51ff6f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/b2a16e4460bf98a5ff750cd8bafb4804e0450a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/b2a16e4460bf98a5ff750cd8bafb4804e0450a0c"}]},"branch":"refs/heads/master"},"0ba0efa921e411ea309d22bf90d9b9535f9eea37":{"kind":"REWORK","_number":2,"created":"2021-03-18 11:59:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/781429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/29/781429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/29/781429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/29/781429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/29/781429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/29/781429/2"}}},"commit":{"parents":[{"commit":"351d69df2d5230639083c6b576aa58dfe10c36c2","subject":"Prepare Wallaby M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/351d69df2d5230639083c6b576aa58dfe10c36c2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:57:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:59:48.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nDepends-on: https://review.opendev.org/781428\nChange-Id: Iae9098ac8a42c8371a817993d4450747a51ff6f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0ba0efa921e411ea309d22bf90d9b9535f9eea37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0ba0efa921e411ea309d22bf90d9b9535f9eea37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
