)]}'
{"id":"openstack%2Fpuppet-keystone~794919","triplet_id":"openstack%2Fpuppet-keystone~stable%2Fvictoria~I8c0589bccbac4776720badfdb009b966ac2f7087","project":"openstack/puppet-keystone","branch":"stable/victoria","topic":"bug/1930403-stable/wallaby-stable/victoria","hashtags":[],"change_id":"I8c0589bccbac4776720badfdb009b966ac2f7087","subject":"Set fixture paths for unit tests","status":"MERGED","created":"2021-06-05 00:35:04.000000000","updated":"2021-06-11 16:43:42.000000000","submitted":"2021-06-11 16:43:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794919-bug/1930403-stable/wallaby-stable/victoria","cherry_pick_of_change":794430,"cherry_pick_of_patch_set":3,"meta_rev_id":"b9871e2758864e6ce6fb245dae65f1eb0884a793","_number":794919,"virtual_id_number":794919,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-11 16:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-11 15:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-11 15:04:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-05 01:22:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39901b1e25729b7049834834ee63ab0a3a086f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-05 00:35:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"6acddeceb9e394f8960b5f0bd796273b3fc88a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 01:22:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4e768d352bb440fb64f3c692e08bb7c : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6363519145d647daa7bb5ff537036a74 : SUCCESS in 4m 15s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d65d411f7de4882acd9df5cfafe5696 : SUCCESS in 17m 07s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be478bcb5de74adc88af75a19a831853 : SUCCESS in 12m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49ee49d6e04c4773817053013822666b : FAILURE in 11m 25s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/902275c15df4487396fd1500ba9c0d95 : RETRY_LIMIT in 12m 41s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d224ebe5742412ba3eaae52af5bb037 : SUCCESS in 41m 57s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4f6b85edd804371913ae15a5023c391 : SUCCESS in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af0c594fc0e6e9e04cb34940fc971ed6f069177c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-08 02:04:22.000000000","message":"Topic cherrypick-bug/1930403-stable/wallaby-ew0154imud removed","accounts_in_message":[],"_revision_number":1},{"id":"58c155ef55bc25c2e07884980eec74cfb495da8d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-08 02:04:25.000000000","message":"Topic set to bug/1930403-stable/wallaby-stable/victoria","accounts_in_message":[],"_revision_number":1},{"id":"75b43c520199532eda44b8247bf379ee10dea998","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-11 15:04:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0cc859d7d36ccdea6e04a4059587512dfa94e2ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 15:52:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49787a3b9b914d44ad90ff102f191a5b : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e006391721cb4e01891036288830de94 : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/efa1005fb29a493d9b0ceff42bdbf41c : SUCCESS in 14m 38s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2826ed3f5b6a481b96f540f99932377b : SUCCESS in 12m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/970e1b6f23c8476881f59d0526923fbb : FAILURE in 11m 32s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/247d003ed2df47d683217a90f7d4b6de : SUCCESS in 39m 33s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/213b84e94b144c52889c753c96c9821a : SUCCESS in 45m 04s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6231edcfdefa4cb09c00724e47e989d8 : SUCCESS in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4eee32626a07319af16653d8f68eed2c6416a6d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 15:53:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"517aaff6bb2fc7a00a5b4648cff6ceb5fac5d337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 16:43:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f625fe5c6e2a43c8823b21c3cb9c0e32 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca34effa2b534e8cb81949e97f4b993c : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7037728148a4a1ca2ff9c1903ef3298 : SUCCESS in 14m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6e0331e9c3544b18ad721507e686eb6 : SUCCESS in 10m 18s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/22b8a772070543d882ee6afb9a460596 : SUCCESS in 40m 46s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/2bf4c574592544a291ff6e19de42a1cd : SUCCESS in 41m 02s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73179c4c9eaf4fa881f2892ae9b742ab : SUCCESS in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9871e2758864e6ce6fb245dae65f1eb0884a793","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 16:43:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"441cea0b57bd4a8559f174eafad22502b45fac69","revisions":{"441cea0b57bd4a8559f174eafad22502b45fac69":{"kind":"REWORK","_number":1,"created":"2021-06-05 00:35:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/794919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/19/794919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/19/794919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/19/794919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/19/794919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/19/794919/1"}}},"commit":{"parents":[{"commit":"3f55c34811b6738cf2c353f11d19c0a3beab0a83","subject":"Add TLS options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3f55c34811b6738cf2c353f11d19c0a3beab0a83"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-02 05:03:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 00:35:04.000000000","tz":0},"subject":"Set fixture paths for unit tests","message":"Set fixture paths for unit tests\n\nThis change defines manifest_dir and module_path expicitly in unit\ntests so that modules installed under fixtures directory is properly\nloaded.\n\nCloses-Bug: #1930403\nChange-Id: I8c0589bccbac4776720badfdb009b966ac2f7087\n(cherry picked from commit c0f33c2806eefc79fdc82a1eccb67a4c3b9ad68c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/441cea0b57bd4a8559f174eafad22502b45fac69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/441cea0b57bd4a8559f174eafad22502b45fac69"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
