)]}'
{"id":"openstack%2Fpuppet-keystone~817911","triplet_id":"openstack%2Fpuppet-keystone~master~I8779d07152b540def0981620cd1c707fe32cd7b0","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-17 10:32:42.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I8779d07152b540def0981620cd1c707fe32cd7b0","subject":"Fix validation of memcache_security_strategy","status":"MERGED","created":"2021-11-15 08:22:43.000000000","updated":"2021-11-17 14:54:33.000000000","submitted":"2021-11-17 14:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817911","meta_rev_id":"ec3f30bc325d6184a6d610862b3c4bb5de59f0f0","_number":817911,"virtual_id_number":817911,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-17 14:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-11-15 15:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-11-17 08:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-11-17 08:09:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-15 10:16:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-15 15:39:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-11-17 08:09:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"db5a7ec81c8c7cd66363ab2e7a804b2094b20379","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-15 08:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33b397e5fcbd3b5bc2b1dbcdb576a4d8192c7f0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-15 08:24:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1532b333e1601e0b92b69c19b629e916cd94b8b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 10:16:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff1cab192ba344d495e64c0909182138 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7992aea3343548baa2f79e2eaf630052 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b00b3b8efad14aa19f192fbaa7d7091c : SUCCESS in 4m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f07d146d2f94e8f94f1c452d98ea2a5 : FAILURE in 14m 35s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef97265a08d649229db37a842ca15ca3 : FAILURE in 14m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d5d8864b42e48a1acd65d259c9c5a48 : FAILURE in 13m 59s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5aa148ac7444c4ba5198d918ea5b5e4 : FAILURE in 12m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c7beb24c7b4436098ba3a612b9575d2 : FAILURE in 13m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/475a4de7fe0244edac330f4c2d82a724 : FAILURE in 1h 10m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95b6f5a348a14c47a00e97e8d7450993 : FAILURE in 1h 44m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f5226e81e554cea9c274c48a1b5177c : FAILURE in 32m 59s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1fd1315469347c58cb71a5d62d63d80 : FAILURE in 47m 35s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0a2a0de424d4118894c5e023ff6798f : SUCCESS in 21m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5febb92354843cfb27a7036acb23121 : SUCCESS in 1h 20m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cdbd7d3759247258c748884a8113d84 : SUCCESS in 1h 30m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/11bad303079c45e39fcbbe52de2c3851 : SUCCESS in 1h 21m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4be607d885a64a5bbf69053ac6a3de3c : SUCCESS in 1h 04m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81a2212c8bac4f92a7fe74f785541b9d : SUCCESS in 31m 54s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/59e825ded5104fb989ffece05fcd8f03 : SUCCESS in 39m 25s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8de9fcabb1484c4da674f5e30ecd41c9 : SUCCESS in 33m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38b8308781d64afca7d398cef411171f : SUCCESS in 33m 19s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dffaba8a6af840baa111d7660353f681 : SUCCESS in 1h 46m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c6cc86728b783d24af5953ccd5aa1ccdee1bc37f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-15 10:46:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"681d89b220e3385163134742c659266fbb88beea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-15 11:49:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"062279eb4fcf47f4fa2f920f45e37e27287efa25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:41:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7218cc3452641b4b808b06236470c4d : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97fca82141974c5dbbf7000d1de0dd84 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a347439babf24748b8d7cbc3e89455be : SUCCESS in 4m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/28d9eae513b542cdbc84dd012e5fd357 : SUCCESS in 15m 49s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/47396c33874b4061a0329c27b487a7e6 : SUCCESS in 15m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54f49d590ea64a64b84a6d96adec3cd8 : SUCCESS in 12m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbea5418233d476e8e9305f0951353a4 : SUCCESS in 13m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65c3d6ae1b9c4343947e52d8918e5a04 : SUCCESS in 12m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a4a21c0d6504190bfd909e323680ee1 : FAILURE in 28m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c714d41f032e4fbca676bad6fe787c5d : FAILURE in 1h 06m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38ac94fdf2cf488eb6d04db5cbd1ac59 : FAILURE in 58m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22f6b887306446c6934664173d33b3ce : FAILURE in 54m 22s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8485b918f574e6cbffd0e07cc5d370d : SUCCESS in 21m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc8f370ef3fb498ab0a5bc1f825e7b8c : SUCCESS in 1h 16m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bef5aa949f9c42a4913283a45c9b3c46 : SUCCESS in 1h 29m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbf3b9cdcf60462fb21680fc309da711 : SUCCESS in 1h 28m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e67cdd84f814c14a47b165d61546723 : SUCCESS in 1h 03m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdf66479930741b4844cb133dc0bc239 : SUCCESS in 31m 36s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/238bc78b47ee4b4c972d676ce47e8e8d : SUCCESS in 39m 25s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaa80405562b49aa890a6f34007213f7 : SUCCESS in 32m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb5f4a2e312f4e168c3db0591701a691 : SUCCESS in 31m 27s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/533fe95c09fa4fec9d783e305718f6b1 : SUCCESS in 1h 46m 49s","accounts_in_message":[],"_revision_number":4},{"id":"803e1c00c596bb6777c2b15181464b086a257e66","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-15 15:39:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e76f5a419322fee4498cfcbe356e5b55bd82a02a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-17 08:09:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"10225dbb027c75bc5e6809f0d0272f29dddeeaf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 08:11:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5be1db00a3212859b098467404a1cd4c27e730ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 10:05:20.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2eb8eaeae8a42f3b15016a35225287a : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dfc9f44667847f4ad084502798a9f2a : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffc08cab28024aeca4909be9b1402582 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a34e6db7ff7847dc933131f82542e2db : SUCCESS in 13m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee833d24450a4f889145cd68887ce456 : SUCCESS in 13m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0d8569724c040b083952150fec2a2a8 : SUCCESS in 11m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38e718dc4a2c49288b63347bac04ab29 : SUCCESS in 11m 02s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/92878f078ec340aface6c5046ed0fec8 : SUCCESS in 1h 12m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/24652eb868c2467289484f91d7d0a746 : SUCCESS in 1h 30m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e06ed9db12f64dfba3a8a10e0454400e : SUCCESS in 1h 14m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fed2cb56d9be46dea4197dc1eb720245 : SUCCESS in 1h 03m 40s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5108d013c4b9495c94f27a547e156dcd : SUCCESS in 37m 35s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c41aea7c5654e638248f9733f61950a : SUCCESS in 31m 25s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/418029d5b79f40e7a210929bf778fe03 : FAILURE in 1h 50m 30s","accounts_in_message":[],"_revision_number":4},{"id":"f06424df3498e1b1e12cd6587b371820521bfeb6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-17 10:32:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c83ba30d685f84931c6e9e003a1a08f664829c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 12:34:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57a23d6dcf11413b9473dfde4198aa3a : SUCCESS in 3m 25s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a626ce4b1e347a7995c7a1ba23318ec : SUCCESS in 4m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8819680ecd946279e1a9201c7b14bf8 : SUCCESS in 3m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa56b290dc494788923dd3ba4a001d28 : SUCCESS in 14m 33s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dacf6101792941dda08b57ba00c54f91 : SUCCESS in 15m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6863298aac2747ab8d109bebec81da7b : SUCCESS in 12m 11s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c830a8ede1246b0ae3f506d17364063 : SUCCESS in 12m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27bf22b66806489e8a5fdb12c3fbedd9 : SUCCESS in 11m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4daa84421e084a478eabbd34237a7550 : FAILURE in 36m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/342026f871004e7ba675020a6d7e2d93 : FAILURE in 1h 30m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f9ac9f44114460fa469107c9ae233bb : FAILURE in 35m 28s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a81eef350aa3473d8137418cf60d4bb1 : FAILURE in 27m 46s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcb801fd50d447e48af8265889a54596 : SUCCESS in 23m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0852f31fad37446e8967eed44c6dcf04 : SUCCESS in 1h 18m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4dd31c2dbfc541bfa6eaf581aa8bab77 : SUCCESS in 1h 21m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/735280f2edfc424bae97afb34df4b4f6 : SUCCESS in 1h 21m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ff567dc43f545a38359cca93b871cec : SUCCESS in 1h 00m 58s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50689b1930ed481eb707f17a78d029e1 : SUCCESS in 33m 05s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4ef1455376a4be59636be2069206e4e : SUCCESS in 47m 28s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8c720b8c9664483b8ec82c8990dcb0f : SUCCESS in 33m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1ccd63698bf49bc9d01253f7d27ee7a : SUCCESS in 37m 45s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7a4bc7c0383c459789a3fe82d111fb7f : SUCCESS in 1h 59m 10s","accounts_in_message":[],"_revision_number":4},{"id":"e91bb7af7edb3a2190b8e4830b7e09704402d099","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 12:36:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"14e93889ff5747f987d4f40887c6c12cedc8910c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 14:54:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15f5ae4e66de424c9dbd6862032eb007 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/832ca125f42d4931a69079d3427a0372 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef0d85e64f95433cb39a1ead8f5cca29 : SUCCESS in 4m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/538c8e86d3824df4bbf89799f73dbfff : SUCCESS in 14m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/217d152ebeca4929b1301eedbc224e2f : SUCCESS in 14m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d684a60bc67484ead6738235bf5cf8f : SUCCESS in 11m 41s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cf31091b32e4a23889f517707a0a033 : SUCCESS in 15m 52s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/14ecb7c697ef43d99ee49e593d075ab6 : SUCCESS in 1h 28m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b73f73a321b240c19e28215694946e7e : SUCCESS in 1h 40m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/69d6235d76e5475f8addae9c137b84e5 : SUCCESS in 1h 13m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/13dda1e49cd440c9b5767b59b6919d07 : SUCCESS in 1h 12m 41s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5344242d7378488ea575923ef5fc349a : SUCCESS in 40m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/744eb468abcc43a8ba00c3bf56f10e7a : SUCCESS in 31m 28s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6e17357c5073456a8c4ab9a725d04e0b : SUCCESS in 2h 14m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ec3f30bc325d6184a6d610862b3c4bb5de59f0f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 14:54:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"38d31574742d4b043279b2a8128eda417d6d662b","revisions":{"5601c368388fc539432da7322ec3cf741a7eef4d":{"kind":"REWORK","_number":1,"created":"2021-11-15 08:22:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/817911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/11/817911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/1"}}},"commit":{"parents":[{"commit":"8c671983c7c56332fd80a099516823c8cce7542d","subject":"Merge \"Add missing dependency on puppetlabs-concat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8c671983c7c56332fd80a099516823c8cce7542d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-15 08:19:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-15 08:22:34.000000000","tz":540},"subject":"Fix validation of memcache_security_strategy","message":"Fix validation of memcache_security_strategy\n\nCurrently the memcache_security_strategy parameter can be set to only\nMAC or ENCRYPT, but None should be also acceptable. In addition,\nthe parameter ignores case thus we should not require a specific\ncase pattern.\n\nChange-Id: I8779d07152b540def0981620cd1c707fe32cd7b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5601c368388fc539432da7322ec3cf741a7eef4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5601c368388fc539432da7322ec3cf741a7eef4d"}]},"branch":"refs/heads/master"},"5e67660a2ef31c9cab3a113f85aa192923ff6f86":{"kind":"REWORK","_number":2,"created":"2021-11-15 08:24:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/817911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/11/817911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/2"}}},"commit":{"parents":[{"commit":"8c671983c7c56332fd80a099516823c8cce7542d","subject":"Merge \"Add missing dependency on puppetlabs-concat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8c671983c7c56332fd80a099516823c8cce7542d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-15 08:19:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-15 08:24:31.000000000","tz":540},"subject":"Fix validation of memcache_security_strategy","message":"Fix validation of memcache_security_strategy\n\nCurrently the memcache_security_strategy parameter can be set to only\nMAC or ENCRYPT, but None should be also acceptable. In addition,\nthe parameter ignores case thus we should not require a specific\ncase pattern.\n\nChange-Id: I8779d07152b540def0981620cd1c707fe32cd7b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5e67660a2ef31c9cab3a113f85aa192923ff6f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5e67660a2ef31c9cab3a113f85aa192923ff6f86"}]},"branch":"refs/heads/master"},"241973881d4da1538120910a61fc7a9379f052be":{"kind":"REWORK","_number":3,"created":"2021-11-15 10:46:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/817911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/11/817911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/3"}}},"commit":{"parents":[{"commit":"8c671983c7c56332fd80a099516823c8cce7542d","subject":"Merge \"Add missing dependency on puppetlabs-concat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8c671983c7c56332fd80a099516823c8cce7542d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-15 08:19:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-15 10:45:58.000000000","tz":540},"subject":"Fix validation of memcache_security_strategy","message":"Fix validation of memcache_security_strategy\n\nCurrently the memcache_security_strategy parameter can be set to only\nMAC or ENCRYPT, but None should be also acceptable. In addition,\nthe parameter ignores case thus we should not require a specific\ncase pattern.\n\nChange-Id: I8779d07152b540def0981620cd1c707fe32cd7b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/241973881d4da1538120910a61fc7a9379f052be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/241973881d4da1538120910a61fc7a9379f052be"}]},"branch":"refs/heads/master"},"38d31574742d4b043279b2a8128eda417d6d662b":{"kind":"REWORK","_number":4,"created":"2021-11-15 11:49:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/817911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/11/817911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/11/817911/4"}}},"commit":{"parents":[{"commit":"8c671983c7c56332fd80a099516823c8cce7542d","subject":"Merge \"Add missing dependency on puppetlabs-concat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8c671983c7c56332fd80a099516823c8cce7542d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-15 08:19:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-15 11:49:20.000000000","tz":540},"subject":"Fix validation of memcache_security_strategy","message":"Fix validation of memcache_security_strategy\n\nCurrently the memcache_security_strategy parameter can be set to only\nMAC or ENCRYPT, but None should be also acceptable. In addition,\nthe parameter ignores case thus we should not require a specific\ncase pattern.\n\nChange-Id: I8779d07152b540def0981620cd1c707fe32cd7b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/38d31574742d4b043279b2a8128eda417d6d662b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/38d31574742d4b043279b2a8128eda417d6d662b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
