)]}'
{"id":"openstack%2Fpuppet-keystone~818163","triplet_id":"openstack%2Fpuppet-keystone~stable%2Fxena~I1855b83ceb377e8c97c351a0434e2ab994fb0bdc","project":"openstack/puppet-keystone","branch":"stable/xena","topic":"oidcresponsemode","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-23 18:19:06.000000000","reason":"Change was submitted"},"34200":{"account":{"_account_id":34200,"name":"Oskari Lemmelä","email":"oskari@lemmela.net","username":"olemmela"},"last_update":"2021-11-18 15:22:00.000000000","reason":"Oskari Lemmelä replied on the change"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2021-11-23 16:36:53.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1855b83ceb377e8c97c351a0434e2ab994fb0bdc","subject":"OIDC: Add support for setting OIDCResponseMode","status":"MERGED","created":"2021-11-18 00:27:02.000000000","updated":"2021-11-23 18:21:07.000000000","submitted":"2021-11-23 18:19:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818163-oidcresponsemode","cherry_pick_of_change":815826,"cherry_pick_of_patch_set":3,"meta_rev_id":"8145c70c1a3268031d24f2534b6032af17421756","_number":818163,"virtual_id_number":818163,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34200,"name":"Oskari Lemmelä","email":"oskari@lemmela.net","username":"olemmela"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-23 18:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":34200,"name":"Oskari Lemmelä","email":"oskari@lemmela.net","username":"olemmela"},"all":[{"value":1,"date":"2021-11-18 15:22:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34200,"name":"Oskari Lemmelä","email":"oskari@lemmela.net","username":"olemmela"},{"value":2,"date":"2021-11-18 22:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-11-23 16:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":34200,"name":"Oskari Lemmelä","email":"oskari@lemmela.net","username":"olemmela"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-11-23 16:36:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34200,"name":"Oskari Lemmelä","email":"oskari@lemmela.net","username":"olemmela"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 00:27:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":34200,"name":"Oskari Lemmelä","email":"oskari@lemmela.net","username":"olemmela"},"state":"REVIEWER"},{"updated":"2021-11-18 02:17:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 22:30:29.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-23 15:29:46.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d864a4a4891fa3e380c12f0faf8e840ba8244619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 00:27:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"72f7c2ea682335291d6c32f08f7b192db142b8ce","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 00:27:54.000000000","message":"Topic oidcresponsemode-stable/xena removed","accounts_in_message":[],"_revision_number":1},{"id":"d74f2322247a6fbd792c3081723f4022104f90f6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 00:27:56.000000000","message":"Topic set to oidcresponsemode","accounts_in_message":[],"_revision_number":1},{"id":"7ce7d5654134f875ef7ce2758ad7fe4af5630c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 02:17:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be562f5ed99e4480b707925cca5ad138 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91d8602409204a8a8dd9c33df968257c : SUCCESS in 4m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73427b789d8f42969daaf883fc1b3938 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c569963c52984e43baa67f65493be344 : SUCCESS in 12m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c23e85fbad64a40bcb7385c2fca89a8 : SUCCESS in 17m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09872e8b11824098a6464be9b853a893 : SUCCESS in 11m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2ec49ccfda644a0bad6564bdc0a7843 : SUCCESS in 13m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79f5a36630b5405da8949ab8647b6bcc : SUCCESS in 11m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89c709e6d01540de9339c04c8a395ac4 : FAILURE in 42m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af66b68f990747c9afcd95964133a48d : FAILURE in 1h 45m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0dbcb267e044a3d9b0c0e4943290b50 : FAILURE in 33m 57s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27a4d44ad5634628a57b6b5bed73e737 : FAILURE in 1h 24m 32s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/83428c0b87684498ace646105c1a8c07 : SUCCESS in 23m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/499b79d5f055485cb0e076620d86994e : SUCCESS in 1h 09m 51s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a18c03ac0749497fb86b2b2cc847d03a : SUCCESS in 1h 13m 49s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4a224c723ee4c15996808d7973a9e01 : SUCCESS in 1h 19m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/077927d902ed4930b2bbdb1c299720a2 : SUCCESS in 1h 00m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a8b384042bf440c82bcfb12a57ef6ad : SUCCESS in 32m 47s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3e2a21e1890451eaee913bde244b93f : SUCCESS in 39m 56s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6e34d3942f940679f5ec98c39549b5e : SUCCESS in 43m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e874908e0d6458b9b1f0adaaa04299e : SUCCESS in 33m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a74fba4bcc34229b58a3ff77e46848d : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ada81c8af9cd3418fdfd31a9826b660d692d6668","author":{"_account_id":34200,"name":"Oskari Lemmelä","email":"oskari@lemmela.net","username":"olemmela"},"date":"2021-11-18 15:22:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a3ed5373f8e5a82ede22c3b87db9e9b613a9c5d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-18 22:30:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"109f78708ef2af2b8b513d9634b54e75b8927287","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-23 16:36:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36f271bced993efdab522ab4253b9921c6b813fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 16:40:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcc9721d06c0803eed8d99864de2402f1342e93e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 18:19:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cce148272c21425b8e4232094c77c156 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/340c8aff700941e1899149ef504e952f : SUCCESS in 4m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21ca02875f384d40b3534187d9845730 : SUCCESS in 5m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/26fe970608bd4fc099e2ff2334a6f3dd : SUCCESS in 18m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/724371393480432781235f082da08c49 : SUCCESS in 21m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1371d83ea65047a0a8f4afa96577adda : SUCCESS in 14m 43s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/951299108fb646e0869e2f0bf1753caf : SUCCESS in 16m 06s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6782e29ea0ba4176bd7ae5fba462e888 : SUCCESS in 1h 14m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/60b382bcbc984385a435f8e52b16f374 : SUCCESS in 1h 33m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3aec48c515f7451681716c4e3ee6de32 : SUCCESS in 1h 16m 35s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/96f00f3835d64d4a96faee4f2f3db102 : SUCCESS in 1h 02m 39s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/818cd4ca6a1548b1a2452ba6c32d8b53 : SUCCESS in 43m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77f525919e774b9081842a1c62b9e93e : SUCCESS in 37m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdd0ceb8e68d47c9bb3c11654b945d8a : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e6c001482abf47d56a539d48e268180404459634","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 18:19:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8145c70c1a3268031d24f2534b6032af17421756","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 18:21:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bd5fd77c8574b70ae75b89894b0b207 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b391b409dd9642bb3bf095549e325111291f3bf","revisions":{"2b391b409dd9642bb3bf095549e325111291f3bf":{"kind":"REWORK","_number":1,"created":"2021-11-18 00:27:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/818163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/63/818163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/63/818163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/63/818163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/63/818163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/63/818163/1"}}},"commit":{"parents":[{"commit":"cd8172d7e2bdfd399470c7cd4d759b72f2f34ba6","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/cd8172d7e2bdfd399470c7cd4d759b72f2f34ba6"}]}],"author":{"name":"Oskari Lemmela","email":"oskari@lemmela.net","date":"2021-10-28 07:42:37.000000000","tz":180},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-18 00:27:02.000000000","tz":0},"subject":"OIDC: Add support for setting OIDCResponseMode","message":"OIDC: Add support for setting OIDCResponseMode\n\nDefault response modes fragment or query are using URL encoding which\nlimits maximum token size. Response mode form_post does not have limits\nfor token size.\n\nThis change allows to define response mode.\n\nSigned-off-by: Oskari Lemmela \u003coskari@lemmela.net\u003e\nChange-Id: I1855b83ceb377e8c97c351a0434e2ab994fb0bdc\n(cherry picked from commit 94fbafd684b04d28a601419cfe10d1e1fbe8b0b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/2b391b409dd9642bb3bf095549e325111291f3bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/2b391b409dd9642bb3bf095549e325111291f3bf"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
