)]}'
{"id":"openstack%2Fpuppet-keystone~818168","triplet_id":"openstack%2Fpuppet-keystone~stable%2Ftrain~Id9dfbab25b7a064e5fee6d8bf4fe6a171fba9b15","project":"openstack/puppet-keystone","branch":"stable/train","topic":"openidc.pp_UPDATE","attention_set":{},"removed_from_attention_set":{"32417":{"account":{"_account_id":32417,"name":"Alessandro Italiano","email":"alessandro.italiano@ba.infn.it","username":"aleita"},"last_update":"2021-11-23 19:12:32.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-23 19:12:32.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2021-11-23 16:37:11.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id9dfbab25b7a064e5fee6d8bf4fe6a171fba9b15","subject":"Unset memcached_servers by default","status":"MERGED","created":"2021-11-18 00:33:13.000000000","updated":"2021-11-23 19:12:32.000000000","submitted":"2021-11-23 19:12:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818168-openidc.pp_UPDATE","cherry_pick_of_change":818167,"cherry_pick_of_patch_set":1,"meta_rev_id":"a9eca7fccc6da6ae3a986bb5e1de075e71b9a7f3","_number":818168,"virtual_id_number":818168,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-23 19:12:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-18 22:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-11-23 16:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-11-23 16:37:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32417,"name":"Alessandro Italiano","email":"alessandro.italiano@ba.infn.it","username":"aleita"}],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32417,"name":"Alessandro Italiano","email":"alessandro.italiano@ba.infn.it","username":"aleita"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 00:33:13.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32417,"name":"Alessandro Italiano","email":"alessandro.italiano@ba.infn.it","username":"aleita"},"state":"REVIEWER"},{"updated":"2021-11-18 02:29:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 22:31:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-23 15:31:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"4534ac8cde3aafa733c8ca622571324a1ce3ac6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 00:33:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"4b164b2e928220dccd2e5749fee0cf19395a7382","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 00:33:21.000000000","message":"Topic openidc.pp_UPDATE-stable/victoria-stable/ussuri-stable/train removed","accounts_in_message":[],"_revision_number":1},{"id":"58839d909ac874c4139eae3fbb4fc016a631be4c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 00:33:25.000000000","message":"Topic set to openidc.pp_UPDATE","accounts_in_message":[],"_revision_number":1},{"id":"f4002dd51298a9d2f9aa1b25dfe13929f37398e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 02:29:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6984e0518e0342749d62694df4b0631f : SUCCESS in 5m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8fcf3e52a4644d284908aa3919946c4 : SUCCESS in 5m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61beffe1c9074aa7bdf0a877c87ff6fd : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/acda5a95635c4259a94b7e00b8999302 : SUCCESS in 26m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f2a4c57ff30f457a840cb42c08a2c848 : SUCCESS in 15m 50s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/5baed30d3aa74a9d8569ae6f7228a292 : SUCCESS in 14m 11s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/254b6ca3d2664e2b939d4a746acafb6c : SUCCESS in 13m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96b605e9ac8d44c1a071b25bea690f60 : FAILURE in 13m 16s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/013931f389e443239c9ec33d05d5221f : SUCCESS in 1h 09m 08s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/29c288f535f44cb3a93be54b5401aa4e : SUCCESS in 1h 20m 16s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a2ef1d5402b44d329c74b16a7fdbbaa7 : SUCCESS in 1h 00m 21s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/d046a4681d0f446baaf4d70713cda308 : SUCCESS in 59m 26s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fd33d70864184adda4638c513e1500d8 : SUCCESS in 1h 01m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/885f31a9d677454eac1146aa45881a59 : SUCCESS in 51m 03s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/658a17c69b60408eb60358de78157203 : SUCCESS in 41m 00s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/565b6d2e74044673aa86939bbe7c6db5 : SUCCESS in 35m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d4963f0247a94b30a60e4772042dae45 : SUCCESS in 56m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f5015a0c57984154aa1613997bbf2430 : SUCCESS in 1h 07m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f958fff70b444bf1ad9722598eb053ef : SUCCESS in 1h 06m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/31be1183e5f74fb794bc0d1d2854de42 : SUCCESS in 42m 52s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ee169ad3c0b9429fa32fd48fd90ab695 : SUCCESS in 41m 06s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f26e698a56b4994a53377a5dca6e906 : FAILURE in 8m 23s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8cd2de7e5eb643ba9f0cafb4a6235177 : SUCCESS in 1h 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ece048e8ded72f64389617272bf38757030d1892","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-18 22:31:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a81501aa7447b0dbb68bb41307102d372c77577d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-23 16:37:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54f6822215b0841293a52ae36b3716476920d704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 16:40:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1740a0a63fd23fff8b130aada868a4ac287e63c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 19:12:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2b1880f909a4812b3797aace184163a : SUCCESS in 6m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc7c83608f584fe9a8ca1c829fff8c7c : SUCCESS in 6m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e94b3971ebf4425099117786bcabddf3 : SUCCESS in 7m 03s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bb8cf8e7ab1143959e36853f5fac04d2 : SUCCESS in 26m 32s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/729ce471993e492eb73560b500e99c0e : SUCCESS in 13m 34s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a8978de853f94721a831175c41d53934 : SUCCESS in 59m 28s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/471d8a29fcb849829652ff2995e81438 : SUCCESS in 1h 03m 06s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/aabfb8d7df5b41208eb4b0a880ed5323 : SUCCESS in 1h 02m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/fcc248cff9954208a1a52b19a3941d8b : SUCCESS in 48m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e7ce784ef1d54980a8527d0c714ec63f : SUCCESS in 46m 01s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/848fa552b37a4101aa7f523a753c0778 : SUCCESS in 1h 41m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a9eca7fccc6da6ae3a986bb5e1de075e71b9a7f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 19:12:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"063d524dbf03648a887fbc6bb3566d8ba8e8bcec","revisions":{"063d524dbf03648a887fbc6bb3566d8ba8e8bcec":{"kind":"REWORK","_number":1,"created":"2021-11-18 00:33:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/818168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/68/818168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/68/818168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/68/818168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/68/818168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/68/818168/1"}}},"commit":{"parents":[{"commit":"5ceb1cd689a31d7c784dc27da2da6dd0092c6c1d","subject":"[Train only] Run TripleO jobs on CentOS8 instead of CentOS7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5ceb1cd689a31d7c784dc27da2da6dd0092c6c1d"}]}],"author":{"name":"Alessandro Italiano","email":"alessandro.italiano@ba.infn.it","date":"2020-10-14 14:06:29.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-18 00:33:13.000000000","tz":0},"subject":"Unset memcached_servers by default","message":"Unset memcached_servers by default\n\nThis change make sure that OIDCMemCacheServers is not set when\nthe memcached_servers is not set, to avoid setting empty value.\n\nChange-Id: Id9dfbab25b7a064e5fee6d8bf4fe6a171fba9b15\n(cherry picked from commit 5c1632c100e89437d17ddae40ee9723fcb9e6c67)\n(cherry picked from commit 9cbe5a1bd601b1bdc4342d8c451e42fada582e83)\n(cherry picked from commit 1dd6ad7ccccb80c357cc4beb96a885f383b47f62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/063d524dbf03648a887fbc6bb3566d8ba8e8bcec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/063d524dbf03648a887fbc6bb3566d8ba8e8bcec"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
