)]}'
{"id":"openstack%2Fpuppet-keystone~818711","triplet_id":"openstack%2Fpuppet-keystone~master~I369d2f2babf027b41cf098afcc6d7b898fea3954","project":"openstack/puppet-keystone","branch":"master","topic":"system-scope-creds","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-25 03:53:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I369d2f2babf027b41cf098afcc6d7b898fea3954","subject":"DNM: Enforce scope","status":"ABANDONED","created":"2021-11-22 05:20:47.000000000","updated":"2021-11-25 03:53:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c17044936e9bce12da9c03aa446db13eb72c39ee","_number":818711,"virtual_id_number":818711,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-24 13:11:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-22 07:00:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee1d43f19e1a8fb807f88c920bbb35398bda866c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-22 05:20:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39ed93cbde1641adca0c8ccf1bcabd7d73b15dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 07:00:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f6ee9c0c10e4ca497399a8de0d04c1f : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1c7ee90125349708f52b9c578133311 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68b5b6b8c493437b91c598dbf4466bb6 : SUCCESS in 3m 49s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/415b12d7a3884149a85d96e056126bc8 : SUCCESS in 12m 38s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7072ed6fda4448168068bc7e6f37496d : SUCCESS in 15m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cce03354408a44fbbb066b66ae8ec0ab : SUCCESS in 11m 15s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a3eefd9d7b94e2b87b3f40cbdaab34f : SUCCESS in 12m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b769227acf24c58b07acbde29c39586 : SUCCESS in 12m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a64be47f65741af8cb48a6abd1cb90e : FAILURE in 1h 00m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a40ea1e94bb408297c5e19d817ebe2b : FAILURE in 36m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/215d67b4f2d04c4d9427ecfd4120134a : FAILURE in 53m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b20dbac15f7407a92cc44f412a367ef : FAILURE in 25m 57s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/da714559d1994fbdbacae82957533c5f : FAILURE in 31m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc328d560a254ba7a19eb79dda189980 : FAILURE in 55m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2c2bf72cfca48139db3d7f8aceec3a0 : FAILURE in 54m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa24206375aa4ceebdae96c1320cc6a3 : FAILURE in 57m 46s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fb375eebc5a4d3d9a1e2ed19ec19a9a : FAILURE in 36m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21a069f56fc141ffae1f31da61772570 : SUCCESS in 34m 15s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3391eb3e270146d08cf55b2a0ee022ac : SUCCESS in 50m 32s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9945072db9994e41b2f5099d26d2466b : SUCCESS in 34m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d21b8d47f51443f84bd2bb77c46d2e4 : SUCCESS in 35m 06s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/17f3defe32064b46abe38d4010f2c09c : FAILURE in 1h 37m 01s","accounts_in_message":[],"_revision_number":1},{"id":"febdb3375cbbd2992121051de077859095acda3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-24 11:06:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"46b9fefed06771d7e78a581e6315ce71db6d831d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 13:11:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d83364be5154d0c97a3a828478158be : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8c9cd8672f94df1b630400bc26767db : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daf1a751d0cb482ca4a4e2b5b7f93060 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8864ec0e1a8f4959ad6f4528664d7610 : SUCCESS in 19m 01s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/21263ddb0653482a9f25bd587fe79a40 : SUCCESS in 13m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea3dcc571bc3423e85e635d35bc9f403 : SUCCESS in 11m 20s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93fea181f11c43e1ab28d4f1c09a13bd : SUCCESS in 12m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a565f57b2b143629284f0f9f830c402 : SUCCESS in 12m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f8b200ef57f46b2a90ff30fa3785fe1 : FAILURE in 28m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bf3e7852e144e70af66d59f74e5e15b : FAILURE in 37m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c06921ff9d54086adf23f25aa857044 : FAILURE in 27m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05373c95a23c40abb0c65899798435e3 : FAILURE in 25m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6144fe65d2df4081ba1288a283458f03 : FAILURE in 26m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d533bc3aa9884bef9eacdc1799fca8d7 : FAILURE in 1h 03m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4168e95a204404aa4bacfca52d0ac2b : FAILURE in 1h 07m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/387d6f00b6b34dac8e315df3bdb33f11 : FAILURE in 1h 04m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/260ce48712c048a8a8057baaf5a523e7 : FAILURE in 47m 10s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d181bd86b00c4d4bb928b0cd2ee60104 : SUCCESS in 39m 28s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/227cd1d8c49240f385d7d47dfbf9b8be : SUCCESS in 42m 11s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7a809f2da4a4628bd7e14e21854f248 : SUCCESS in 42m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8605e67080f44179159aedde8d8e398 : SUCCESS in 34m 10s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a5893317313b451f8775cc4b356738db : FAILURE in 1h 56m 48s","accounts_in_message":[],"_revision_number":2},{"id":"c17044936e9bce12da9c03aa446db13eb72c39ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 03:53:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"403b949af8361b33409f42c268110fda75437859","revisions":{"5701379128d7177541d3fccc74c71bb7fb958221":{"kind":"REWORK","_number":1,"created":"2021-11-22 05:20:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/818711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/11/818711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/818711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/818711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/818711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/11/818711/1"}}},"commit":{"parents":[{"commit":"fd7a62e29a9c2d30c5bdef459c189048ab4cf3a3","subject":"Use system scope credentials to request keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fd7a62e29a9c2d30c5bdef459c189048ab4cf3a3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-22 05:20:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-22 05:20:32.000000000","tz":540},"subject":"DNM: Enforce scope","message":"DNM: Enforce scope\n\nChange-Id: I369d2f2babf027b41cf098afcc6d7b898fea3954\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5701379128d7177541d3fccc74c71bb7fb958221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5701379128d7177541d3fccc74c71bb7fb958221"}]},"branch":"refs/heads/master"},"403b949af8361b33409f42c268110fda75437859":{"kind":"REWORK","_number":2,"created":"2021-11-24 11:06:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/818711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/11/818711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/818711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/818711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/11/818711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/11/818711/2"}}},"commit":{"parents":[{"commit":"fd7a62e29a9c2d30c5bdef459c189048ab4cf3a3","subject":"Use system scope credentials to request keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fd7a62e29a9c2d30c5bdef459c189048ab4cf3a3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-22 05:20:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 11:06:09.000000000","tz":0},"subject":"DNM: Enforce scope","message":"DNM: Enforce scope\n\nChange-Id: I369d2f2babf027b41cf098afcc6d7b898fea3954\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/403b949af8361b33409f42c268110fda75437859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/403b949af8361b33409f42c268110fda75437859"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
