)]}'
{"id":"openstack%2Fpuppet-keystone~819512","triplet_id":"openstack%2Fpuppet-keystone~master~I858b446effb95505fec207d93c314d591d2aa34a","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-07 23:09:51.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I858b446effb95505fec207d93c314d591d2aa34a","subject":"Clean up deprecated ignore_default_tenant","status":"MERGED","created":"2021-11-28 13:08:44.000000000","updated":"2021-12-08 04:44:14.000000000","submitted":"2021-12-08 04:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819512","meta_rev_id":"6fdccd77276124e6b3a43a03a1034544d8de8442","_number":819512,"virtual_id_number":819512,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 04:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-29 08:24:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-07 15:55:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-07 15:55:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-28 15:30:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-29 08:24:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-07 15:55:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e02a4c8bd4b57274cbef0ea1d4c2ccc63b611c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-28 13:08:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e538d0c86831eee70115a84d5605df819ca4c2ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-28 15:30:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cbb0b5e3f7840a0acf5c6f2281bffef : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0509e44ebe44ddf8562faa0770e179a : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f58b2ec1370c4e4b93c5aa539a085c5e : SUCCESS in 3m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a23cfabdafa845e7b95360e3150613fd : SUCCESS in 15m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f511c593a0bc4b8594c9e96cc0c5b096 : SUCCESS in 12m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2627eaeedebc4cfab768bd0fa01d9d45 : SUCCESS in 12m 19s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d5d827f53a34cecba675212691a62bf : SUCCESS in 10m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dab2737aa51418490257f80520c4069 : SUCCESS in 10m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9e8813ec9fc46c6825470f92213119d : FAILURE in 35m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e29f6ced33654d2ba10357e46d04c73c : FAILURE in 1h 13m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d7d3a52937d4c4c966c336745430e4a : FAILURE in 30m 02s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9840912f5b054673978b94c2c4c6f224 : FAILURE in 55m 42s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1839bf4056543bfac287aef49957d5d : SUCCESS in 23m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5af844a54dfa4e3e9545992f06ff09f9 : SUCCESS in 57m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e8770bc85934ed58cc5cecacbee5082 : SUCCESS in 1h 48m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf396cee092c4133afe26e06141ae04d : SUCCESS in 1h 05m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1288a2af4578406fa6acfd08ecd5ab1f : SUCCESS in 1h 08m 20s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32e5be1e81694b48872abcace2115502 : SUCCESS in 16m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa4fe8a738074b159bd130a339357724 : SUCCESS in 46m 39s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e111c2b0232845b29c34b6c2d7cc5040 : SUCCESS in 47m 17s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/efa38d5a101f4678bff11dc7c8107a22 : SUCCESS in 52m 28s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/174099dc9608405cbbb04e190dc9595f : SUCCESS in 53m 23s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e0ef1ed9d7d4478b7cd0bda4bdd6b6a : SUCCESS in 1h 01m 18s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/178c6abc97d14a02ae8204682d4e0fad : SUCCESS in 35m 19s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b353fc8756b448b97edb824127fbfba : SUCCESS in 37m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e88deffd5aea45a29f84fb9779fb37c4 : SUCCESS in 30m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01b915352ab344319622ab1be2bbdad3 : SUCCESS in 3m 11s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e3bd1c3de93f437487a88584ccf3cda2 : SUCCESS in 2h 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"779650c307581d75f90c59763c066edd1609ecb0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-29 08:24:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f348c62b4928626eaa119a3ddf31282662839f98","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 15:55:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe59471bc2fcea093dfa7a2ec88031012285989f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 23:09:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0878ed7c0928281b58a8c93846ffee2dc709f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:02:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c56f3d8ff06049759221c65ddc6946eb : SUCCESS in 4m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff11179434d94b868e8a588cec93a74b : SUCCESS in 4m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/570e9c1cb7f54217808bebf95514d779 : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/54380aa342e245d9bcc1bb1e683437d0 : SUCCESS in 16m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/503b7e2f1a3046d6b05fcabbe56fb047 : SUCCESS in 18m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b77f868bc98478a808b58d52a9d03b7 : SUCCESS in 12m 29s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9b7fa4b62444ea99b85744df51c237d : SUCCESS in 16m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7f8f8ae752f47139d74749675e18b1e : SUCCESS in 13m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/126b870089d3453d930ef953050e2f69 : FAILURE in 36m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cecee06f71854dcda6cff035223f8ea0 : FAILURE in 35m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f24ffb5251204ee9be8ab9ed32d08ac9 : FAILURE in 36m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/649f9c33d4004cb6b22b7fc7ce3d7c5e : FAILURE in 1h 01m 12s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a003ba9b98c040788d3b6314a426ceba : SUCCESS in 33m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ad3cc13b4f14f828108a1f73a3ec536 : SUCCESS in 1h 21m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d09874f131446868825e355c49d545b : SUCCESS in 1h 31m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/00b704e7dcc144fc9fb3a0db9345219b : SUCCESS in 1h 21m 24s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/edadb810bd084d5f8951e91f72ab49bb : SUCCESS in 1h 03m 18s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14f7a6834c2c49469c618f0c941be442 : SUCCESS in 25m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6e08909a2664a4fa9f4feef08ed9ea6 : SUCCESS in 52m 33s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/65b7a31c25954a8a92d5b71b440c0bd3 : SUCCESS in 1h 08m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a24aacc66b74551ad6abb75029a2939 : SUCCESS in 58m 41s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/159787618c3b4e47aa78c06a802573f1 : SUCCESS in 47m 28s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c50e6d41afe14f3db64a683ff78e1bf3 : SUCCESS in 29m 24s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b57f97c4b72744b09061b7848aa8ff9e : SUCCESS in 39m 13s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53f62387c4f749018cc51bde5a7905de : SUCCESS in 30m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99c5f33bf432469db9c7a715694c3e1b : SUCCESS in 42m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47b0f75c727049aa91577838a894192d : SUCCESS in 5m 53s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/556dd92bc1a64b56b56283405b7c7eda : SUCCESS in 2h 25m 03s","accounts_in_message":[],"_revision_number":1},{"id":"604baf26a86cb1c97315605370208b26657fb876","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:13:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71675a666c7df69ff317defc07c9da5900e3ed30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 04:42:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/023bf3dfb9094fa68c08eb5dd518e1eb : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/586d0d2a991440c592417d3b099c7a0d : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf4078bc45b84a5e87b01dfb830d6bc9 : SUCCESS in 5m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/36c44f65a9f2411fa5163716d96b6656 : SUCCESS in 14m 37s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c1c42cf2ceb47efaa7a62780480d19f : SUCCESS in 16m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a125aeb9db9140e39d62849a1c44bd14 : SUCCESS in 13m 05s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77c427f7e4414e70bd0e42d118943821 : SUCCESS in 12m 48s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5410db8d1704dd6aa0af36c30148b22 : SUCCESS in 1h 12m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/95a5a75fc7aa401b90cea33d66487aca : SUCCESS in 1h 37m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ffcbfb5ec98e47868e1395d0c9009355 : SUCCESS in 1h 18m 17s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e6b35107b1243a58a10c25b48e6c0f7 : SUCCESS in 1h 15m 52s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d16fd1ee95724ba1a7a6b1bad0c2fe5c : SUCCESS in 46m 26s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b81cd8e9c2b41279640f5749a5dae85 : SUCCESS in 34m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33531c95b1ae4691b1502ae8b7d99787 : SUCCESS in 4m 46s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bcc4ccb838614efb910f4b1eee7e0ec2 : SUCCESS in 2h 01m 17s","accounts_in_message":[],"_revision_number":1},{"id":"aa67cb1135de6967608c6d9429cc659cf7ce6e0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 04:42:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6fdccd77276124e6b3a43a03a1034544d8de8442","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 04:44:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3e1bd47b3174397980e4aa0a707ba87 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26827ea1d7b3af14d500ac35daa2df541b5ec169","revisions":{"26827ea1d7b3af14d500ac35daa2df541b5ec169":{"kind":"REWORK","_number":1,"created":"2021-11-28 13:08:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/819512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/12/819512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/819512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/819512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/12/819512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/12/819512/1"}}},"commit":{"parents":[{"commit":"102f8d7fca26e2015fd0eda56807d17a39feb076","subject":"Merge \"Fix validation of memcache_security_strategy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/102f8d7fca26e2015fd0eda56807d17a39feb076"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-28 13:07:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-28 13:07:50.000000000","tz":540},"subject":"Clean up deprecated ignore_default_tenant","message":"Clean up deprecated ignore_default_tenant\n\n... because this parameter was deprecated during Victoria cycle[1] and\nhas had no effect since then.\n\n[1] 72796c30b24fe4af0b154c14eba88f8c57d5f74d\n\nChange-Id: I858b446effb95505fec207d93c314d591d2aa34a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/26827ea1d7b3af14d500ac35daa2df541b5ec169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/26827ea1d7b3af14d500ac35daa2df541b5ec169"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
