)]}'
{"id":"openstack%2Fpuppet-keystone~834297","triplet_id":"openstack%2Fpuppet-keystone~stable%2Fxena~I9a1f452c3797ccbda25865ca78f1e3a24103d070","project":"openstack/puppet-keystone","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-30 00:29:10.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I9a1f452c3797ccbda25865ca78f1e3a24103d070","subject":"Ensure inet6: formatting for a comma-separated list string","status":"MERGED","created":"2022-03-18 13:45:53.000000000","updated":"2022-03-30 04:18:12.000000000","submitted":"2022-03-30 04:18:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834297","cherry_pick_of_change":834180,"cherry_pick_of_patch_set":4,"meta_rev_id":"bbdf6eb707bb5f8713fde7558cb9dc267773564b","_number":834297,"virtual_id_number":834297,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-30 04:18:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-03-18 21:25:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-03-28 20:25:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-03-28 20:25:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-18 15:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-18 21:25:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-03-28 20:25:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"942e26113c3b84dcb61877b65d9306e7e475404d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-18 13:45:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d362775525c9bcc64c5a925afe06690e56297fa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:23:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/baed68af262c44248c4de9496c784246 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecb78d237f384ff58694e6889ad56556 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51a02f4dc5844482b81c6ed794ef9785 : SUCCESS in 4m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/43bf13ac58f04110ab82954e206a8b4d : SUCCESS in 14m 51s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/61b2d55f25f74b578fbb8d72d6025d3d : SUCCESS in 14m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27ec67d632594f0faf2b10a974106f47 : SUCCESS in 12m 53s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e0dfbd0b3704ab68b97e7e67572fab4 : SUCCESS in 11m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11d424a41f4f452e89ab09978fd623ba : SUCCESS in 12m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e8bee8e85bd4b058b1527fcda6b5075 : FAILURE in 56m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/002b21404fae4dd081fc084864de80c4 : FAILURE in 37m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed51369758ea43eaae09db32396cbb3d : FAILURE in 49m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/584a512a4cb24e4b9dc4fd86d46e4480 : FAILURE in 31m 04s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b54a9fce849e4b698db14d7ebbc93a8d : SUCCESS in 23m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf2c95832c11466b80a18e753ffd68c3 : SUCCESS in 1h 14m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a5792d6dccd4524a42fc2aa49a990a9 : SUCCESS in 1h 31m 58s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e343f55f2b5c49549c9f9c9b94940dad : SUCCESS in 1h 16m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b03a6b64a7c4b0d8c8bafa5ec769225 : SUCCESS in 1h 04m 32s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55e0af93cffd435ba8f4e9e15f41c8b3 : SUCCESS in 31m 23s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b85e3957696492fa29a14ecc408ff9c : SUCCESS in 39m 06s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2d453a0b4bd44349645794d41d69534 : SUCCESS in 32m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37117c6da2d64029bd74704768c19be9 : SUCCESS in 32m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2936032bf001452d25e5b9acaa41aac4edd1c43","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-18 21:25:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"649518d744c1c0daeb200f58e81f537ac1005fd8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-28 20:25:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"faf75e00266b7fe3486bdd01df83ede0a3a2c49f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 22:12:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3a2127da6baf40394218e6061941cb71111ec58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 23:43:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc74b990839e46729c457809e1c3cf5b : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ec6062a05564fa197042f7402217373 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3994430452e54fda8e784840ea046a1e : SUCCESS in 4m 29s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d9598ebf6c542c8a508f25222ad48b9 : SUCCESS in 15m 45s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/05e0c5862bf54731872fde17bcd06279 : SUCCESS in 15m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5116268401d547a4a55eba0ef4054f0c : SUCCESS in 13m 04s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/003e87d55f1a4daabc280bfcb0863567 : SUCCESS in 12m 06s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5de7ca9b96b4b4c943b04112d2913e9 : FAILURE in 14m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/166f8e5180734e3990f61586fd2a25a8 : FAILURE in 14m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e0470b8e01a4f6a92693599f5e89d05 : FAILURE in 14m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a6b148f1a3843a6b25404aff98f1977 : FAILURE in 13m 22s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/316269b0122c4371bde3a3a08c5abb5f : FAILURE in 37m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b467b319888e47239304c6b7dae4f0c7 : SUCCESS in 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ab6c7b0f6acbe8e4ae196f48b71d133ae3651cd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-30 00:29:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"808b0b4c0f96955dca2bde7168e4b068a670900f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 02:31:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6bba53861174769969f1c5705fa5e3c : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ceac3c9d5484284b6872840f43fcc41 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ccb2e2ea8bb40b5a9bb9c2b152a572f : SUCCESS in 4m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4cf82e3e7a414c6c964a3319632df8ee : SUCCESS in 16m 22s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b660db732ef49ffba5aa55d276a1158 : SUCCESS in 14m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e912f9c2be343d2af5a6b98ae76f900 : SUCCESS in 13m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/705a3344e17a4f13933ec5256f38f555 : SUCCESS in 13m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8823574d1834214994caff5cdfed04b : SUCCESS in 12m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c0391096dab462eb36029c1e30368c0 : FAILURE in 39m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c62c478383e4cdfbc34715a0f50dcc6 : FAILURE in 39m 22s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2059e8f2017e4c968af285dcee1f1da3 : FAILURE in 1h 05m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a4bc87c33b54d4192f19a6721174d1e : FAILURE in 52m 47s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/06edceabdc7b4016983084c5247339bc : SUCCESS in 28m 18s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/85d0a72e9ec5460689b2a086aeec5601 : SUCCESS in 1h 32m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7114a442f9184f8d9a81822d07837f45 : SUCCESS in 1h 55m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed6f3b5dda4c48afac35a3ad7544bbca : SUCCESS in 1h 41m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/99c926c681ab4f1cb6ab10e14b5d7bf5 : SUCCESS in 1h 15m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/938028d871ff488ab40e3a5d194cb183 : SUCCESS in 35m 44s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/08ad957c14864b48bbbf6fdc794a58dd : SUCCESS in 50m 35s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5d4b257824348bf929fedc81291018a : SUCCESS in 35m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41013e0f33344d5cb2263b1a837dc29b : SUCCESS in 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06df9f2b98ab3fbe9215cc1a605daeb66af22c18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 02:32:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9bd414eb24501d901601ad3daed7bedb2ef393d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 04:18:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a04dce6b9d8845e3b124b2ab81a2e43d : SUCCESS in 6m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/535a311a7864459bbf224861884b6952 : SUCCESS in 6m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89e91f2cd49f4e4b88ff38096018f600 : SUCCESS in 7m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c59e0e197de468e818a4370e575bc4f : SUCCESS in 19m 22s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d176d2fd4004e3a9fc0c87409b5a881 : SUCCESS in 22m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b47d3d2625ab4ec8b498ead191ca4dc6 : SUCCESS in 14m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16d4a6b9e0f048abb707cfab030bead2 : SUCCESS in 19m 30s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d936128974ef44379637a67b318e09bf : SUCCESS in 1h 24m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8da63b5073224b7cbdfee9a11aef2da4 : SUCCESS in 1h 31m 52s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e62c97ded2554117a6c2fe93176b4acf : SUCCESS in 1h 39m 29s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7238115d6adb42fe8ddc88f4bf7c84e4 : SUCCESS in 1h 06m 30s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/91bce221321d4abfbf0476bdc425211c : SUCCESS in 48m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10ac221076b54f2d84454c689729cc6d : SUCCESS in 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbdf6eb707bb5f8713fde7558cb9dc267773564b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 04:18:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed8ab833d8cb489aef0ee9d5f24da8840c6c0ffa","revisions":{"ed8ab833d8cb489aef0ee9d5f24da8840c6c0ffa":{"kind":"REWORK","_number":1,"created":"2022-03-18 13:45:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/834297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/97/834297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/97/834297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/97/834297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/97/834297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/97/834297/1"}}},"commit":{"parents":[{"commit":"bcc37d771ba95b44e3f0bf7dbc51610355a20326","subject":"Use systemctl to restart the keystone service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/bcc37d771ba95b44e3f0bf7dbc51610355a20326"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-17 14:51:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-18 13:45:53.000000000","tz":0},"subject":"Ensure inet6: formatting for a comma-separated list string","message":"Ensure inet6: formatting for a comma-separated list string\n\nCurrently the inet6_prefix function recognizes a comma-separated list\nas a single string, thus the format is not applied.\nThis change ensures the value is converted to an array and the format\nconversion is properly enforced.\n\nChange-Id: I9a1f452c3797ccbda25865ca78f1e3a24103d070\n(cherry picked from commit c2525c6e63c236b470bd935496875545f91e2b26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/ed8ab833d8cb489aef0ee9d5f24da8840c6c0ffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/ed8ab833d8cb489aef0ee9d5f24da8840c6c0ffa"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
