)]}'
{"id":"openstack%2Fpuppet-keystone~859378","triplet_id":"openstack%2Fpuppet-keystone~master~I1fcefad64225ea9917605d451237967edb4843ed","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-11 04:52:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1fcefad64225ea9917605d451237967edb4843ed","subject":"openidc/mellon: Use static Location for protected endpoints","status":"MERGED","created":"2022-09-27 07:02:56.000000000","updated":"2022-10-11 04:52:06.000000000","submitted":"2022-10-11 04:52:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859378","meta_rev_id":"210774831892cb6e10747132d2436a6042ed3a38","_number":859378,"virtual_id_number":859378,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-11 04:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-09-28 14:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-10-05 09:55:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-10-11 02:47:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-27 07:24:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-10-05 09:55:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-27 07:24:15.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2022-09-27 08:49:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-28 14:40:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-10-05 09:55:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-10-11 02:47:51.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"c5b721231d28fb0a9573c7066bc687f6033e1fad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-27 07:02:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bb936f7e12e7aeff41235f9af91c180aa4da095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-27 07:04:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d9d7aba9c01fd56f8264566d7e8eb6cb91eb657e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-27 07:10:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6d53d90d690419d752d499602c26182ddf8df313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-27 07:19:39.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fb6a02d7dd8d1c200b16a98a0a9d755f3f4c202f","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-09-27 07:24:15.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cf6dadde9e2551bd5575ce9fd22b6a55bbf1810a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 08:49:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/136fa68c2bb1442aa5c44254b633eeab\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a805947e8cea4fd28a8444c9018fdb7f : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5152c8c186ed4e3a84f6e61224c5e459 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1b27dcfbe184922b0527aef15704518 : SUCCESS in 5m 21s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b1a368594cc4709bb78bdbb7acab327 : SUCCESS in 26m 04s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7673f75bfb204c0dab5938f8831c548d : SUCCESS in 20m 48s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1c3e462a9964034b8c314dfe73d3039 : SUCCESS in 21m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0ec2b7b124249a3b2754c5c6d63d8e5 : SUCCESS in 20m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b165a533e1a44583a1a18a4db9b5a07d : SUCCESS in 55m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d8b99d2edf84def82779e41df4be8f2 : SUCCESS in 1h 10m 28s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05f0de1be2354b2a8514a50718ed4ee4 : SUCCESS in 35m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/729e7a58beb04296b8db7b9eb245877a : SUCCESS in 38m 19s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20541323866d4afa803eb502792ac80d : FAILURE in 48m 55s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4fccb7c5685492fb2ca34b87a19a7e2 : SUCCESS in 23m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/92a4735404b6440ca6d1ce4cef538136 : SUCCESS in 1h 13m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ba92c4316a64969970e980681fd195e : SUCCESS in 1h 07m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a9bda2e1a384f74acbaccd3762423ae : SUCCESS in 1h 06m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bddd1d256c74e489cfe2dd52ca5c8ce : SUCCESS in 1h 05m 30s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fc561f730524fbc8f97492f7ad265d7 : SUCCESS in 48m 27s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44086bb7f9234e2ba30f23887895dcd5 : SUCCESS in 32m 28s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9a5c394f41e472f8ce20a5368e7ed7e : SUCCESS in 39m 39s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b75709d4fc940b7b45ccecff0c16b0d : SUCCESS in 32m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13f410c2dabe4b95b4462a500fb3c0d3 : SUCCESS in 43m 38s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1022360469d74d1faf7f6834b8e5b1fd : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":4},{"id":"652d30a31a58df1deaf4e02a59b9750dc677458d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-28 14:40:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"74de1efd0ae7770662cb6628fa721b509564392f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-05 09:55:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ffbfa0de6c495c42416e8e3b682e6138a3e40f14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 09:57:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5dc3ba40b6b1fcfa38bb032b980aefb34ce60fc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 12:53:21.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf1f465526044ab99f3d15b674d5bf96\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e1c455364aa47048baf619d0711425e : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f38cc92c875432590a2072d25e86eeb : SUCCESS in 3m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b97e7da9d1743d5b9f3936f0bbc52f8 : SUCCESS in 4m 00s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2100bfca73de48eb810dbc77d6611fe9 : SUCCESS in 14m 08s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3a39898c3a040b1927c1331c6c803c2 : SUCCESS in 11m 22s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dab978b0b7204be385b0bd69d842c632 : SUCCESS in 10m 53s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6382ec66f5114d699ede4ae8f50bc28f : SUCCESS in 1h 03m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b79da83d711a425492b76c963d7cb899 : SUCCESS in 1h 01m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/48475a4c4b3b47bb8b5aa6d68d985fae : SUCCESS in 1h 15m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b023e7b8928b45cfbe05b1a3694d355c : SUCCESS in 59m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6eea15ca5f946cd81882e7cdc25e54b : SUCCESS in 58m 58s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa1cfa034f104e5aaec108421f20b0be : SUCCESS in 44m 22s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0881050a2c06408f88b4c105e202de20 : FAILURE in 2h 54m 39s","accounts_in_message":[],"_revision_number":4},{"id":"b9bf598efdc635ecf3c373f0fe45b1c319129533","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-06 00:09:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"26ad6301dd3799b52852e37ccd70e74d78832445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 03:07:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d135b2a6a13847c4a31fc6034691ad3a\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68911d775e9043f185a3406aa8f9761b : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16c246aae4d747d7a4471c74305f1fb1 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/297cd8d367ef4ae58a6944f7661c462f : SUCCESS in 3m 32s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/315a1a1ac88c4358ab885a1039d5e055 : SUCCESS in 13m 09s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e3a239d5d95478987d31e3f86e4a9b8 : SUCCESS in 10m 39s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a25278dac494cf0a4e8af8020e83ef0 : SUCCESS in 10m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3decbc5d38194e1ea2354c92948ae016 : SUCCESS in 10m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/475fb26341554c8d9cf6d902456e77a4 : SUCCESS in 43m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bd794d4b6364076832512797ddb4dca : SUCCESS in 58m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a39397f46b6745caa7d4dffc22644b54 : SUCCESS in 36m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1824b33f419b47a99d6f1bb4d5139122 : SUCCESS in 43m 59s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5159d92824674b0eae199c6da154d735 : SUCCESS in 42m 16s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bc46ece48af4a99ad6a103983888695 : SUCCESS in 20m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b2d348374114eaa9e932a8363d012c2 : SUCCESS in 1h 05m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/84c6430eb1ae42858efbad2ac4fefa19 : SUCCESS in 1h 14m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/323eff5c4cb947c5911b50a6701d3c53 : SUCCESS in 1h 08m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/559b499b0bea47c987f2e678e3b61870 : SUCCESS in 58m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/292e83dcb0784e4f901bf865d4d24317 : SUCCESS in 55m 32s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c590df0af4741d390684c83a30c53f4 : SUCCESS in 39m 17s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8c696cc0aec49958d3bfe7c8b21ed18 : SUCCESS in 41m 45s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82a0d381aa164f7486fb775609bbabbf : SUCCESS in 35m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef37d15479614d83b314c22d043c119e : SUCCESS in 35m 41s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aec08507f1d64835bb8ae9d6c12d23db : FAILURE in 2h 54m 49s","accounts_in_message":[],"_revision_number":4},{"id":"9a00eb1795afc3c187c50f8f7630e60f7c983386","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-06 08:16:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2a98587753dae19bebebc2238bac083e9b79e33b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 11:21:44.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f9091a27b1940818d13bd07971f5ad2\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f157e0e764e944d0bb8889fca1ce310c : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f491c4d49ca43d6b7c4bb7af1cf9827 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd9c0e4b69e9496b8d0d0ec71c24eaa4 : SUCCESS in 3m 51s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4afc18e833d642919619dd2874dfbac6 : SUCCESS in 15m 56s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3285be38b3f34f2490326cdbc27808d4 : SUCCESS in 10m 59s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a8606570e2e481d8f9ca0166e9148ab : SUCCESS in 18m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/451f2fe957304a46817ca0e9b982583a : SUCCESS in 10m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edf166caf731468dad9c895873d0cb41 : SUCCESS in 44m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c146da1c858a40999568e0267c080bf6 : SUCCESS in 56m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e702b2e888d40d693372b226cf0d66f : SUCCESS in 43m 56s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e43b4584e63a497b8bc812ab6511bc2a : SUCCESS in 38m 10s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78017e44e7bb4576b8eb04e4e93d29f4 : FAILURE in 38m 19s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ffe06c798a8496ba7de87bc4d94c98d : SUCCESS in 22m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3d46184c6044b7c8c9e554070ffbe08 : SUCCESS in 1h 00m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30182b29fe58401cbcc357e2fa0c79b9 : SUCCESS in 1h 09m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f984a0e664a647aebec9758fc25beede : SUCCESS in 1h 05m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a5529a07319426bb2cfa4e8874ba691 : SUCCESS in 48m 58s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1751fc405c33432dace9454ac0d14a24 : SUCCESS in 54m 03s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/776c7875ccb046a3b672d4de56cd54ff : SUCCESS in 34m 01s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c70371bd9c94547981a2f3668b21e86 : SUCCESS in 40m 43s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92168bdff40244cd97cc3bd309d32ecd : SUCCESS in 32m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5eb1180e96f4cdbafa0baa7e6bdba76 : SUCCESS in 34m 56s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/156f471af595426e801d453d7bed984c : FAILURE in 3h 00m 04s","accounts_in_message":[],"_revision_number":4},{"id":"4713745a1f40e4f126ac32994ede447934115138","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-11 01:37:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ce4d665b7b9f5791fabc8e25390eb7381e871ab8","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-11 02:47:51.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6c9bd35b78bb2fbbbb9d8d99848cedf9b3e26f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 03:07:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2fc52fbcc6f4584afab3e997352944f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1129b280f8f2475db957bff1c9a7e14e : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33e1a405063645b5a756ec4fed5d894c : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edd94881681740a88e6e1c23246918e3 : SUCCESS in 3m 07s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a726dae20dfe4068b67d4412558d6acd : SUCCESS in 13m 54s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0a76eb7b6014c44b3846f0d951ecd90 : SUCCESS in 11m 36s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0acb49bfc194804a35a2d08b17ebdb3 : SUCCESS in 12m 10s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/076386a911f349d5b00dbcadb16f375c : SUCCESS in 9m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dea6673974a54aaf829316b69f4d21dc : SUCCESS in 43m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdc3ff16ee6b45b088b8f9e3c24b6a54 : SUCCESS in 1h 01m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c5ecdb049a645e2a1d899236ad43b9e : SUCCESS in 42m 18s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6b67d4a6a3b4e1e843099bd821baa5e : SUCCESS in 46m 36s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57f77d612f464c01bc4403ee921fca1d : SUCCESS in 39m 29s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43105ef00e6645d2a3d6b10d6014a6bf : SUCCESS in 19m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dc2ffe3f4a548379a351d06f4629a0f : SUCCESS in 51m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e0a36af45d64512bf84c87efd07692c : SUCCESS in 1h 22m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/87807cff310f4d0fa81b75f98f4984cc : SUCCESS in 54m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/64faeaa8a58043f4a4d3714c1a8f6f9f : SUCCESS in 54m 57s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9273c2d38ef48c5bba1f1c68856413f : SUCCESS in 48m 38s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0702f5522e6e43a184df332e53bd2913 : SUCCESS in 35m 45s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/30b39bc523834c2a9267c490034d0def : SUCCESS in 36m 46s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6de9a8ae559d4536843f51e4587066f3 : SUCCESS in 43m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/602e392a8d434d0fa243445aac6018c0 : SUCCESS in 35m 49s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3ab128786c854eccb9a51f007da12e9c : SUCCESS in 1h 21m 43s","accounts_in_message":[],"_revision_number":4},{"id":"4b6a8d0e4fe9ebae189cfb554f971b20244902e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 03:09:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"019a72d0578c25b72777212ceca8969e8062e43c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 04:52:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fc212d039224bdca477cac99b6826ff\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a43e4116e304fba86bf47912bd66dc4 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f346df7ad9a4f76b2d4e9d31b699c56 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c02438b703aa42bca7cbe367ba256ff8 : SUCCESS in 3m 29s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/af48fd341f514f4093750f89e33dbbb6 : SUCCESS in 13m 51s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9863fd131f714ef08dcf8808ca6d8eb2 : SUCCESS in 10m 44s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4859b71ea60948b495c54c0dd096e032 : SUCCESS in 10m 04s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bcdcfbc6e39466e925ce50660c3d9ff : SUCCESS in 54m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72da7d0675524131b71f3d6f44b882d1 : SUCCESS in 1h 08m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/09ca6ad417a748e7acf586de4b73f401 : SUCCESS in 1h 04m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/495c23dc61fd4f43b06b49d8d96a18e7 : SUCCESS in 49m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecd6d919964c419581be5dd7b91e70c1 : SUCCESS in 54m 28s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/09154f899ed147dc9f065f9dc4758e31 : SUCCESS in 38m 49s\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/45a89dec27c0498dbc58dffbb6794742 : SUCCESS in 1h 42m 04s","accounts_in_message":[],"_revision_number":4},{"id":"210774831892cb6e10747132d2436a6042ed3a38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 04:52:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fd2ab9f6060b722fb6db610c6d6163ccaeae0c99","revisions":{"d35016a4ef57ad4f6aa9a1d7a90cc5af016aa5d6":{"kind":"REWORK","_number":1,"created":"2022-09-27 07:02:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/859378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/78/859378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/1"}}},"commit":{"parents":[{"commit":"31c42f0dd6cfa659fea5dca224630afd7d0ce426","subject":"Merge \"Allow customizing --date option for trust_flush cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/31c42f0dd6cfa659fea5dca224630afd7d0ce426"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-27 07:01:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-27 07:01:49.000000000","tz":540},"subject":"openidc: Remove unnecessary usage of LocationMatch and regexp","message":"openidc: Remove unnecessary usage of LocationMatch and regexp\n\nThe current usage of LocationMatch and regexp (Location with ~) is not\nrequired because the static paths are configured. We can replace these\nby the bare Location directive.\n\nChange-Id: I1fcefad64225ea9917605d451237967edb4843ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d35016a4ef57ad4f6aa9a1d7a90cc5af016aa5d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d35016a4ef57ad4f6aa9a1d7a90cc5af016aa5d6"}]},"branch":"refs/heads/master"},"dcab666cb84d4e81c2d2513dda4384c83c148056":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-27 07:04:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/859378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/78/859378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/2"}}},"commit":{"parents":[{"commit":"31c42f0dd6cfa659fea5dca224630afd7d0ce426","subject":"Merge \"Allow customizing --date option for trust_flush cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/31c42f0dd6cfa659fea5dca224630afd7d0ce426"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-27 07:01:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-27 07:03:38.000000000","tz":540},"subject":"openidc: Remove unnecessary usage of LocationMatch and regexp","message":"openidc: Remove unnecessary usage of LocationMatch and regexp\n\nThe current usage of LocationMatch and regexp (Location with ~) is not\nrequired because the static paths are configured. We can replace these\nby the bare Location directive, as is described in the Keystone doc[1].\n\n[1] https://docs.openstack.org/keystone/latest/admin/federation/configure_federation.html#id5\n\nChange-Id: I1fcefad64225ea9917605d451237967edb4843ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/dcab666cb84d4e81c2d2513dda4384c83c148056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/dcab666cb84d4e81c2d2513dda4384c83c148056"}]},"branch":"refs/heads/master"},"fe22d10586a59b2570100f13d95f2b0c90b36d88":{"kind":"REWORK","_number":3,"created":"2022-09-27 07:10:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/859378/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/78/859378/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/3"}}},"commit":{"parents":[{"commit":"31c42f0dd6cfa659fea5dca224630afd7d0ce426","subject":"Merge \"Allow customizing --date option for trust_flush cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/31c42f0dd6cfa659fea5dca224630afd7d0ce426"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-27 07:01:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-27 07:09:52.000000000","tz":540},"subject":"openidc/mellon: Use static Location for protected endpoints","message":"openidc/mellon: Use static Location for protected endpoints\n\n... instead of using LocationMatch or Location with regexp(~)\nunnecessarily. Usage of the Location directive is described in\nthe Keystone admin guide[1].\n\n[1] https://docs.openstack.org/keystone/latest/admin/federation/configure_federation.html#id5\n\nChange-Id: I1fcefad64225ea9917605d451237967edb4843ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fe22d10586a59b2570100f13d95f2b0c90b36d88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fe22d10586a59b2570100f13d95f2b0c90b36d88"}]},"branch":"refs/heads/master"},"fd2ab9f6060b722fb6db610c6d6163ccaeae0c99":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-09-27 07:19:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/859378/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/78/859378/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/78/859378/4"}}},"commit":{"parents":[{"commit":"31c42f0dd6cfa659fea5dca224630afd7d0ce426","subject":"Merge \"Allow customizing --date option for trust_flush cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/31c42f0dd6cfa659fea5dca224630afd7d0ce426"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-27 07:01:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-27 07:19:39.000000000","tz":0},"subject":"openidc/mellon: Use static Location for protected endpoints","message":"openidc/mellon: Use static Location for protected endpoints\n\n... instead of using LocationMatch or Location with regexp(~)\nunnecessarily. Usage of the Location directive is described in\nthe Keystone admin guide[1].\n\n[1] https://docs.openstack.org/keystone/latest/admin/federation/configure_federation.html\n\nChange-Id: I1fcefad64225ea9917605d451237967edb4843ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fd2ab9f6060b722fb6db610c6d6163ccaeae0c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/fd2ab9f6060b722fb6db610c6d6163ccaeae0c99"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
